)]}'
{"id":"openstack%2Frequirements~795479","triplet_id":"openstack%2Frequirements~master~I12f7ef29499b132d07fdc2ba8d3201e5edf16f0d","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I12f7ef29499b132d07fdc2ba8d3201e5edf16f0d","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-06-09 06:22:08.000000000","updated":"2021-06-09 10:53:17.000000000","submitted":"2021-06-09 10:51:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795479-openstack/requirements/constraints/noclob","meta_rev_id":"8592d54524e150daec1703c8980fccaa8a9cfdb7","_number":795479,"virtual_id_number":795479,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-09 10:51:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-09 06:34:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-06-09 06:34:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-09 06:33:35.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-06-09 06:59:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-09 08:52:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c488831b168d339e98f45ed1f173785fdd12e1ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-06-09 06:22:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16f878c1c6f646e80f38d61ba42a18283110e787","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-09 06:33:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a3c3c696459fda2fbfaca875f5857fd62a7c1448","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-09 06:33:48.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"61f868ce86b050f205efe14aacf9c6bbd2e6b351","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-09 06:34:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2e0a78d29c960883e11c3c9419fe5ab01d7acdee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 06:59:00.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d91dd89bdcd6465db4fb45b6b06389e5 : FAILURE in 23m 30s","accounts_in_message":[],"_revision_number":2},{"id":"f12599505054dbe968d79495169616587bd24d76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 08:52:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/014649230dba4b799ab8bd90bd4d5552 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a547ea6636da4d99a42e7f25f4b10d63 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7ac6174133f476989a80ed58ec6fd5d : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7526d668ff554ae692edd796bcf2a7b4 : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69280ffc4d304fdaae4d3094fa2fcc27 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a3261511da5642cb9bec15affb7128e6 : SUCCESS in 5m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2c628b41a223410a8be8f67f851490f6 : SUCCESS in 5m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2811d22669ae416e91de1fb10c7f7efa : SUCCESS in 11m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8db8e81730c3438bbc5c4468d10aab99 : SUCCESS in 12m 49s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/368d9592e31a4524bee6e04c4b5a2299 : SUCCESS in 12m 52s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a54af85acb81499a99593556bac5423a : SUCCESS in 6m 05s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6a0988bd022a4cafa74712150654539c : SUCCESS in 10m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/84ecd1e8e2514b4b9da2f78415cd37b7 : SUCCESS in 9m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/dc7da6540f184534959a234c35a3f161 : SUCCESS in 10m 41s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b18d02fbafbf4c3dbb673d552612597b : SUCCESS in 15m 15s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7218a9c0742742a09e4d907cccab23a8 : SUCCESS in 9m 40s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9eb41be4bbc242259076ee6d386b1fb9 : SUCCESS in 6m 23s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9ae3933d3a604629b1373d285fe75072 : SUCCESS in 24m 01s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8dfa2827947244edac23f816c65b4e40 : SUCCESS in 15m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9f514ea5d44e460d8542361f84383d06 : SUCCESS in 5m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ca4bab87b3d34315a411f93c3f8de880 : SUCCESS in 17m 42s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b53d33e185e642f8833a6ed766567d84 : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a5a1b49d785349c089e05847a202e3e9 : SUCCESS in 8m 00s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0034034d5bfd484db315355c0a66549f : SUCCESS in 11m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94b1c85ca24c430f9b2bc72ff9cc8a62 : SUCCESS in 49m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/46d25d62ec9044ff9208a3664e623151 : SUCCESS in 1h 38m 01s","accounts_in_message":[],"_revision_number":2},{"id":"d3ccd8e480c1730d52183d99aa01e5f195c84885","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 08:53:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"55cb9789c2a0a4e739c01caa494d42b3300ced56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 10:51:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8b121745ecd4e52b2af768835c3aaa0 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8401c0772bc743528aa3be5d2ac48cbb : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e2db2ef88ed40ed9fa96757f1a7401e : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b17f11d825824f56b149fe5fbd550442 : SUCCESS in 5m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/17b14c2ff94f458e94d77ea43927b650 : SUCCESS in 4m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9fe4a73895f543e99ef5b8e4b3346f78 : SUCCESS in 4m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4fb90bc27a4f450180f8d1d8ac58781c : SUCCESS in 12m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6d3986bb9f3f4c7f8fb7fff5dbc28d60 : SUCCESS in 11m 16s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/210980757e08457c98d223580bb975ec : SUCCESS in 11m 59s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5ee6bdcca5c5443bb888b60f3b4726e4 : SUCCESS in 5m 47s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/eedf01715aaf45e9a28650b1179cab07 : SUCCESS in 10m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2fdfdb802deb49ab9a8adc6c13668b24 : SUCCESS in 8m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e06757e1ff544528812da4253a1d5cd7 : SUCCESS in 12m 38s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ae9a215484d34b38863b6717db5a7d27 : SUCCESS in 13m 59s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ea691741dca841cdb92b61cfa6118ca3 : SUCCESS in 9m 54s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ed49fea5115e4abdac25acb22fd4d996 : SUCCESS in 7m 25s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/96abe62af2ad434dab070d8f84ab9afb : SUCCESS in 29m 08s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d0df7bb320bb433582facdfdca799c6f : SUCCESS in 13m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d494e69664ac4c17ad05e8b4a49e5236 : SUCCESS in 5m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/250d363a2d574370ad233b186d9fbef3 : SUCCESS in 20m 19s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f83fcfc4386e4f048bd855ffc4be31e5 : SUCCESS in 6m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/84d474b6658347188eb46f2e9813ed12 : SUCCESS in 6m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0f10a3c1489e49d1b91581904813ddfa : SUCCESS in 12m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b415c32778549c2962db8b462252292 : SUCCESS in 45m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/114228c7af0f4f88accbabaa65d944ad : SUCCESS in 1h 22m 30s","accounts_in_message":[],"_revision_number":2},{"id":"23d627050574e93084dc3bf6fbc0b9d5713c4a1a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 10:51:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8592d54524e150daec1703c8980fccaa8a9cfdb7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 10:53:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bc928f0cbdd432381a863eb117fd68e : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ba5e968ada8a36a011bb609da6bad6e3de908018","revisions":{"8796df40568adc95ba8277a5cd923f8b4eb625c4":{"kind":"REWORK","_number":1,"created":"2021-06-09 06:22:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/79/795479/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/795479/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/795479/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/795479/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/795479/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/795479/1"}}},"commit":{"parents":[{"commit":"748e981ff163732e486e7f0d589e4907fdd52454","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/748e981ff163732e486e7f0d589e4907fdd52454"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-09 06:22:05.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-09 06:22:05.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I12f7ef29499b132d07fdc2ba8d3201e5edf16f0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8796df40568adc95ba8277a5cd923f8b4eb625c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8796df40568adc95ba8277a5cd923f8b4eb625c4"}]},"branch":"refs/heads/master"},"ba5e968ada8a36a011bb609da6bad6e3de908018":{"kind":"REWORK","_number":2,"created":"2021-06-09 06:33:35.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/79/795479/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/795479/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/795479/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/795479/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/795479/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/795479/2"}}},"commit":{"parents":[{"commit":"748e981ff163732e486e7f0d589e4907fdd52454","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/748e981ff163732e486e7f0d589e4907fdd52454"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-09 06:22:05.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-06-09 06:33:29.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I12f7ef29499b132d07fdc2ba8d3201e5edf16f0d\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ba5e968ada8a36a011bb609da6bad6e3de908018"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ba5e968ada8a36a011bb609da6bad6e3de908018"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
