)]}'
{"id":"openstack%2Frequirements~795532","triplet_id":"openstack%2Frequirements~master~I9437e48ad382c6e04a37bfe2448ad1299e72d1af","project":"openstack/requirements","branch":"master","topic":"types-paramiko","hashtags":[],"change_id":"I9437e48ad382c6e04a37bfe2448ad1299e72d1af","subject":"Add \u0027types-paramiko\u0027 for paramiko type hints","status":"MERGED","created":"2021-06-09 12:41:32.000000000","updated":"2021-06-10 08:33:39.000000000","submitted":"2021-06-09 18:42:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795532-types-paramiko","meta_rev_id":"9fd2f03b8260b682bdbd3335865fff5165bc1145","_number":795532,"virtual_id_number":795532,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-09 18:42:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-06-09 14:06:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-06-09 15:34:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2021-06-09 15:08:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":1,"date":"2021-06-09 13:21:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2021-06-09 14:47:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-06-09 15:34:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-09 13:04:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-09 13:08:01.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2021-06-09 13:21:01.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2021-06-09 13:21:05.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-06-09 14:06:50.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-06-09 14:45:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-09 14:47:09.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-06-09 15:34:49.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-06-10 08:33:39.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"3bdc7580d5ed2b204c92847c4578fdaa220a4654","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-06-09 12:41:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0e90e87a8dcf88a5de9f03a82d47428ac08382f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 13:04:07.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/482563086de84a1f95d3dd48b34ef77b : FAILURE in 20m 46s","accounts_in_message":[],"_revision_number":1},{"id":"fb9dcfe5e8fe3bee023ca881f3681818594f1b3d","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-06-09 13:08:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a1f0a6f077b4d8ba089ca461bf0062847b604b63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-06-09 13:09:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c20b4fe77db7872ac374fe69835ec4414c79101d","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-06-09 13:09:53.000000000","message":"Patch Set 1: Code-Review-1\n\nApologies, the nova change that depends on this is still failing with the following:\n\nhttps://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_2bf/795533/2/check/requirements-check/2bf67e2/job-output.txt\n\nERROR: Package \u0027types-paramiko\u0027 is used in test-requirements.txt but not in lower-constraints.txt","accounts_in_message":[],"_revision_number":1},{"id":"d24ee56cfa89929840154a9d4ea29ee687bfaaaf","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-06-09 13:21:01.000000000","message":"Patch Set 2:\n\nYep since mypy\u003d\u003d0.901 I see the same error locally in nova in the pep8 tox target.","accounts_in_message":[],"_revision_number":2},{"id":"d25d19b8cc01fecaf293a732507e94af99c1d28f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-06-09 13:21:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4508367862735ad0bf2a579f866f33a72382af23","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-06-09 13:25:06.000000000","message":"Patch Set 2:\n\nmypy 0.9 did a change in the type stubbing https://mypy-lang.blogspot.com/2021/05/the-upcoming-switch-to-modular-typeshed.html that is why we see the problem now.","accounts_in_message":[],"_revision_number":2},{"id":"64437779dd0569e0088480a15ec498c3783eab19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 13:30:21.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b21846e3161f45e8bfddeb1f629e12b5 : FAILURE in 18m 21s","accounts_in_message":[],"_revision_number":2},{"id":"85afe1d0e1940e253f47a01db544ec106e281357","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-06-09 14:06:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"23f20f74deac4a286c8aecb4332d0badbc018c31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 14:45:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c5b2708dd424aea9534f357b2a68870 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c131878029ae4a81af5020e9a148be24 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/710ad6c92cfd4118a555b4e59b3a28d4 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3792849a9f04a2e9faaca9fb0f22491 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e9193c347b545218b0b4ecfef332b55 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4d053f3121fa4cde9094b10be74d1494 : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/767f9222c2134a47a4f508cbe1a0b336 : SUCCESS in 4m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/db23dfac6434449d8b61757dfe3e837a : SUCCESS in 10m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9748911e6de4438caddcfb98e85a5051 : SUCCESS in 11m 58s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6cf912e3a9cf471199db71599a9cf815 : SUCCESS in 11m 09s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/748e6d04ed9542efab549513efdc389a : SUCCESS in 5m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8a3462f707754b45b795f376e81653c6 : SUCCESS in 11m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d08e4af3f8dc4766837cebba962a463a : SUCCESS in 14m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ab7195e302ca4fccaf907a4396d77bed : SUCCESS in 11m 41s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0ab484f86ae840d79d8e1997966ce40a : SUCCESS in 13m 52s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/684192b13f484f58ade5f7c2b0076e54 : SUCCESS in 10m 10s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/48eeb71c7a5c40a18216159523cb6a8c : SUCCESS in 6m 17s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/263e4867febd481e91b372a6a4cc2850 : SUCCESS in 27m 02s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b857688ed18f4c78ae92b2509d972343 : SUCCESS in 13m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6d20e9bb30824a10be2d10bbf9da3ed5 : SUCCESS in 4m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6aa65c87a50749649395b293a515a34f : SUCCESS in 18m 56s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/33f2f6b8344149c092260ce9f93853fa : SUCCESS in 6m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7aece9e07f3f4bc0affa16227a5bb812 : SUCCESS in 7m 59s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e199b73a2af548c2bd2ee040478212c1 : SUCCESS in 11m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3a1003756594e3eac241eaeed66b6aa : SUCCESS in 56m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/77d13ac1cff04ca89069901898552ce9 : SUCCESS in 1h 30m 47s","accounts_in_message":[],"_revision_number":2},{"id":"440874fa053653dedcd209fb5a3d565eaeb6a16e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-06-09 14:47:09.000000000","message":"Patch Set 2: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":2},{"id":"3793119eaf4644ba8a1f300174b20bc7fab8ab7f","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-06-09 15:08:29.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM now thanks!","accounts_in_message":[],"_revision_number":2},{"id":"e3c1b199e6508c9f04c59167e08badef55eb76b1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-09 15:34:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0629a960010e0d541c282931b7d465e552383054","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 15:35:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f1577cd0a550f2b48761eacaee8bc22f15b96dbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 18:42:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd5e9905cdc447ef93ffe528761d6849 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ae01d0cae4a4d2b8b03759922aef1db : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0384b4d689d846fa8339bed5176a96f3 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3bb1bbf3115d46ceba727925e061c2cb : SUCCESS in 5m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c45bc90624e14cd38891a21dfabfa136 : SUCCESS in 4m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ceb432b08fd34ff09afe697cc09b1c34 : SUCCESS in 4m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/91948a819b51452b8acd72fe84eb657e : SUCCESS in 11m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/083838e288a44f908fe74d88f12c7349 : SUCCESS in 10m 44s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/80edde46e3db4d4ab49c1eefec4d2bea : SUCCESS in 10m 56s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b559996645094a41b9b354e7eaff0812 : SUCCESS in 6m 30s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/14473cb185d44fd7a78ec0a82b81222d : SUCCESS in 11m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f31f1671f50749089be1742d7fe7d132 : SUCCESS in 10m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c028822ae26a4c71ab0e5e54abe0e0c0 : SUCCESS in 11m 45s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6dc9ab4607da4c74807f01a2a420ed64 : SUCCESS in 14m 02s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/179e706db4d34be1a1e0e9ea21267203 : SUCCESS in 9m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b92216e20087402c9c963e37c604c64f : SUCCESS in 7m 03s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/04d0aa056afb4059b15997909f933bde : SUCCESS in 26m 29s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c6e8e810d4ef470c8f0c156cb1a688c4 : SUCCESS in 13m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d9bc727ab5634c7c8c91941b1cc63375 : SUCCESS in 4m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/17b2fa0ae5ba4deeab16276f2925e2f9 : SUCCESS in 17m 44s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/37b4046ae69b4621b8b61d352a51b15e : SUCCESS in 5m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a7e1a1c4f41d474ba0e01828ed65b690 : SUCCESS in 7m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7d281d8017df4fff8e3f58bf0dc6b410 : SUCCESS in 14m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/295f3eb66094436aaffa44a90db2283e : SUCCESS in 49m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b157c46094ad45568b461c88dfc05857 : SUCCESS in 1h 31m 42s","accounts_in_message":[],"_revision_number":2},{"id":"7da71562ef6085066928a03a1123562a163c99e1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 18:42:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"994548ca2b1c4c283a281c3c87a26fcc09b085d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 18:43:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2780138ee2ab4ac487b436135a7fac5a : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"653b9130b3abb70a5a3285250dbb5e7fe1841ef8","revisions":{"072c25f4c166fbee14a5553579c43f9ac89eae8b":{"kind":"REWORK","_number":1,"created":"2021-06-09 12:41:32.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/32/795532/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/795532/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/795532/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/795532/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/795532/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/795532/1"}}},"commit":{"parents":[{"commit":"ba5e968ada8a36a011bb609da6bad6e3de908018","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ba5e968ada8a36a011bb609da6bad6e3de908018"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-06-09 12:34:29.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-06-09 12:41:11.000000000","tz":60},"subject":"Add \u0027types-paramiko\u0027 for paramiko type hints","message":"Add \u0027types-paramiko\u0027 for paramiko type hints\n\nThis is the split out type hints for paramiko from typeshed [1]. It\u0027s\nnecessary for nova to avoid the following issue:\n\n  error: Library stubs not installed for \"paramiko\" (or incompatible with Python 3.9)\n  note: Hint: \"python3 -m pip install types-paramiko\"\n  note: (or run \"mypy --install-types\" to install all missing stub packages)\n  note: See https://mypy.readthedocs.io/en/stable/running_mypy.html#missing-imports\n\nThis can be resolved by installing the hints package, as the error\nsuggests.\n\n* Is the library actively maintained?\n  Yes. It\u0027s part of typeshed [1].\n\n* Is the library good code?\n  Yes.\n\n* Is the library python 3 compatible?\n  Yes.\n\n* Is the library license compatible?\n  Yes - Apache 2.0.\n\n* Is the library already packaged in the distros we target?\n  No.\n\n* Is the function of this library already covered by other libraries\n  in global-requirements.txt?\n  No.\n\n* Is the library required for OpenStack project or related dev or\n  infrastructure setup?\n  Yes. It\u0027s required by nova, which uses paramiko and type hints.\n\n[1] https://github.com/python/typeshed\n\nChange-Id: I9437e48ad382c6e04a37bfe2448ad1299e72d1af\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/072c25f4c166fbee14a5553579c43f9ac89eae8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/072c25f4c166fbee14a5553579c43f9ac89eae8b"}]},"branch":"refs/heads/master"},"653b9130b3abb70a5a3285250dbb5e7fe1841ef8":{"kind":"REWORK","_number":2,"created":"2021-06-09 13:09:28.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/32/795532/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/795532/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/795532/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/795532/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/795532/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/795532/2"}}},"commit":{"parents":[{"commit":"ba5e968ada8a36a011bb609da6bad6e3de908018","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ba5e968ada8a36a011bb609da6bad6e3de908018"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-06-09 12:34:29.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-06-09 13:09:02.000000000","tz":60},"subject":"Add \u0027types-paramiko\u0027 for paramiko type hints","message":"Add \u0027types-paramiko\u0027 for paramiko type hints\n\nThis is the split out type hints for paramiko from typeshed [1]. It\u0027s\nnecessary for nova to avoid the following issue:\n\n  error: Library stubs not installed for \"paramiko\" (or incompatible with Python 3.9)\n  note: Hint: \"python3 -m pip install types-paramiko\"\n  note: (or run \"mypy --install-types\" to install all missing stub packages)\n  note: See https://mypy.readthedocs.io/en/stable/running_mypy.html#missing-imports\n\nThis can be resolved by installing the hints package, as the error\nsuggests.\n\n* Is the library actively maintained?\n  Yes. It\u0027s part of typeshed [1].\n\n* Is the library good code?\n  Yes.\n\n* Is the library python 3 compatible?\n  Yes.\n\n* Is the library license compatible?\n  Yes - Apache 2.0.\n\n* Is the library already packaged in the distros we target?\n  No.\n\n* Is the function of this library already covered by other libraries\n  in global-requirements.txt?\n  No.\n\n* Is the library required for OpenStack project or related dev or\n  infrastructure setup?\n  Yes. It\u0027s required by nova, which uses paramiko and type hints.\n\n[1] https://github.com/python/typeshed\n\nChange-Id: I9437e48ad382c6e04a37bfe2448ad1299e72d1af\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/653b9130b3abb70a5a3285250dbb5e7fe1841ef8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/653b9130b3abb70a5a3285250dbb5e7fe1841ef8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
