)]}'
{"id":"openstack%2Frequirements~795926","triplet_id":"openstack%2Frequirements~master~I2a240e6284b802bc830418ffafd8fdaa4a4a5780","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I2a240e6284b802bc830418ffafd8fdaa4a4a5780","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-06-11 06:24:42.000000000","updated":"2021-06-16 01:55:26.000000000","submitted":"2021-06-16 01:53:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795926-openstack/requirements/constraints/noclob","meta_rev_id":"5da64182b2efa59daa1916fbd8115f6ce9edf09d","_number":795926,"virtual_id_number":795926,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-16 01:53:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-15 23:01:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-06-15 23:01:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-11 06:39:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-11 08:08:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-15 19:35:50.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"d69b9ec3f488ff067fa03a3a81acdbe9995a07ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-06-11 06:24:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6304edc456fd2c94f99430618edb5ae9d2e9e6e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 06:39:54.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a7ac05fe903848d58f50e0e70d2f638c : FAILURE in 10m 16s","accounts_in_message":[],"_revision_number":1},{"id":"9eaf9136340e0d33bba7c59615c8a7cc6650c7e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 08:08:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6787bd2697e84196b55518dc53ed7f7b : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/499041a768484d0aa99d30ca4c087be1 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c57750cd2dfa42259cbf78b5420d3416 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/273042d87b844279bab06f1822538c4f : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94801884d15c48b7aa1bd80db332450f : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6308d0b5202d4df2ad27a7247b01dd7b : SUCCESS in 5m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cc40f48702ca4f5d94fe4d8c0f9c8330 : SUCCESS in 6m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1a54273cd0ab49c3a931a395de9ccc6e : FAILURE in 7m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c28fcf4e80bc47169b8b276c3ecba6d1 : FAILURE in 27m 08s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7bf4822bc149471aaa4655bb3b401a7a : SUCCESS in 10m 28s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a208e3b15cef4d27be1eaa0eed089e60 : SUCCESS in 6m 41s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b842eafa2b784a2f88ab2d6d5b493139 : SUCCESS in 11m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/324af28cd3fc4bc4991240b1bb1f3efd : FAILURE in 8m 23s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d60d1b146aed4e0fbc4fa631f83c6962 : SUCCESS in 10m 33s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/62b442b7b81d4ed38d0cc42b14fa50f5 : SUCCESS in 15m 30s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2dac41f8ac9b4611becdd697e622d8f7 : SUCCESS in 10m 42s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/3f0bca2de7014caa9f472ea531375172 : SUCCESS in 6m 53s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4f97588d8cd04ec3bac92c1562ff0dbd : FAILURE in 21m 55s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b10150191f4744bdaa31696019f8db47 : FAILURE in 12m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/85e1e993c675489296451a5d54d6e580 : SUCCESS in 7m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cd5faa8f7d4f4d13a4f93d582263e7af : FAILURE in 18m 43s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/223efd5e3ba649d9a4d4651cf09dfd76 : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fb0bd6181cf744c2ac7e88cf58008270 : SUCCESS in 6m 57s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/29d9888eb64f45d59d11ee8fdb23c646 : SUCCESS in 13m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/67eda11799794f31a01cb6fffd8cd6c7 : SUCCESS in 48m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/32377b44a3e14377beae1c7392223728 : SUCCESS in 1h 16m 15s","accounts_in_message":[],"_revision_number":1},{"id":"771cd8771a189b34c8286e917b71334a379f4956","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-06-12 06:24:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"23e6b42b426496a307191b1234b6f2bcdf7237dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-12 06:41:49.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/97d20b93ea8b4081bccdb95568375fa0 : FAILURE in 13m 12s","accounts_in_message":[],"_revision_number":2},{"id":"a0195075d718c09a7cbdf2fa1f2457ab67ed8667","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-12 07:53:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32f2990315fe4d399d4f3214282ca4b1 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/137fda7c96c1442c8b52c65fec94257c : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9a3bce6a3124394bf1ddb4fbcdf6860 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3892d1d0902d47508f99412803a8a767 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b83ef514ac8c48bbb0f233bbd3d4e6b0 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3f1806d237924c5fa7286581bc6cecaa : SUCCESS in 5m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a78e4420ab5840cf941a25a4c5040a9a : SUCCESS in 4m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/01ea61959cbc4103b7d1ea1090c8950b : FAILURE in 6m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9d55634ab89448dcbf3d5f9a3d5a1d7d : FAILURE in 9m 04s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5dcf86c3ab7f4f9ea975bd851bf1430b : SUCCESS in 12m 17s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/99ac1ce377e8462e9eb57b934910c1c3 : SUCCESS in 8m 06s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3a2dab3496f74052893acc479093b1af : SUCCESS in 9m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0ed4372360fb4741841083092e6dffcf : FAILURE in 10m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/683644a2eec54ac1baf5c739d8efe33b : SUCCESS in 10m 59s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e71f6ad38f9c467fac0ba68c49e36466 : SUCCESS in 15m 30s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6609495e68e74111bbbe4963ce4ddaed : SUCCESS in 10m 37s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/df0cb78ea9c44f8485944fd4f0d05f91 : SUCCESS in 7m 50s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/144ffb5f9f3a429baa36a02345d4a54c : FAILURE in 22m 31s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/80ed05d5be27474bb2799eed9d516326 : FAILURE in 14m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/60fa2684d67c4601b28f1ee73b2d51f7 : SUCCESS in 6m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5ed29b5f007941d3ae7b1f7467ae860c : FAILURE in 22m 22s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a71f0387497b4f88a10aaa73c2e97073 : SUCCESS in 6m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fdbea692d899409eb24fe3f6d5243954 : SUCCESS in 9m 09s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a337b766b35646018add73180dcaf539 : SUCCESS in 18m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4bfb85b1f76f4148a9c57d937e431f22 : SUCCESS in 49m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f31d450d495044498dd88becb92021d0 : SUCCESS in 1h 16m 43s","accounts_in_message":[],"_revision_number":2},{"id":"805fcbb957fe97c5a403e50757efaee94a0d398f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-15 19:35:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"481a6a2833b1f053108de86c5dbef7d287e4704b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-15 19:35:57.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":3},{"id":"eddb7a2599429ede6eab5215029cdf15b49b9f10","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 19:57:18.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a24fbcf3755b4b1bb112722d4419f36d : FAILURE in 13m 51s","accounts_in_message":[],"_revision_number":3},{"id":"ac6f5397f7b765a632a286db48d3bae3744873e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 21:19:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/143dc3d6cb0c41d6900a58eb83c654a6 : TIMED_OUT in 33m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed04ece7ae564f409f5fcd87f391dc84 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29ac55ab4c734181b4251eaddb7de42a : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea1e2ec425d7406ba3fd0698658b8a97 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffaf60be691e4d91b260359e4342a987 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bd6a496e1f694cf491b1c267859a50f3 : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e415df3e0b3d4af28989f80042e5d998 : SUCCESS in 4m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ddd679cbd2b74478885655bbec8d942a : SUCCESS in 10m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/40939e04e4ff4909be5dbdb629b27c1a : SUCCESS in 10m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/1715a7b7d2024c5e8148860918281c5d : SUCCESS in 10m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/42c6337b5f5c47fdb725cda3d0e8c568 : SUCCESS in 6m 20s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/01770436636545ed8a68ff05c4f4d858 : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/65ea411875604c7494561c397d658530 : SUCCESS in 7m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ab9ee14bd37147bfab0598655429778c : SUCCESS in 10m 29s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a8cda30557924285934bbb338c7fe8ee : SUCCESS in 17m 05s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/86136fd5205f475b875c4a675ec69640 : SUCCESS in 9m 00s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/941b9cb59171432184cc0dea9a055e5e : SUCCESS in 6m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6760b943d64b488e8355a6b5b7a883ba : SUCCESS in 23m 17s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6c489fda5ce946dbaa6b916b20d1e393 : SUCCESS in 14m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f766b7df94d945858ceb60d69c59d45c : SUCCESS in 3m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d7d586d41f6c438c9bf2b66e2d39bf19 : SUCCESS in 19m 27s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b7eb7cbeca3f4819bc8c12f89a413b65 : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f736b824e7e34bf389c7e62b2259f112 : SUCCESS in 6m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1dc1f69256504e30ae47cc94893ba03a : SUCCESS in 11m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a51054129284181ad04eb98572fbf93 : SUCCESS in 48m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/24aeb0a3ed4c4b0ab66654de0e9432a9 : SUCCESS in 1h 39m 21s","accounts_in_message":[],"_revision_number":3},{"id":"4b4ad97af3264633254bc9c1ebd4be230a4e350f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-15 21:29:07.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"34a21b1e0cfb14c8541fdd6db71a8f1e12bd2f40","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 21:52:44.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f44fd026d9ab4faf84fdf30f86a1ecd4 : FAILURE in 12m 15s","accounts_in_message":[],"_revision_number":3},{"id":"e99f0351740e71415aae4f5bbdc8b6ce1907087e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 22:49:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76e11c92adae47069bb7d707b280e81a : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3955888ae868448ab9592dfe10ca39d9 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5637e291a02b41239b4e0aa05631f540 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64339eec53e64afe935d94481e2a994d : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ed7f2ad28034c41a9f76cff05a91a81 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bb978ec759aa45caa4d19f37645d0b40 : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/79434ec59fe9420fa98c5330f4485e90 : SUCCESS in 3m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/48f9db51341349ab9d91c4e599d1b001 : SUCCESS in 9m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a905069716fe43cfa1936d4409525943 : SUCCESS in 11m 42s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8bd94009513045cbb70b7ac25609c6cf : SUCCESS in 11m 15s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/6bf8e4c8fedb459e9819a40d88d92ce1 : SUCCESS in 5m 46s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3b857698d3924fe1a7913e49d56da932 : SUCCESS in 11m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/247788d01c6645e99ee195b7b1132d4a : SUCCESS in 9m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/31ef48ac96a344d3986d0d80512726d2 : SUCCESS in 9m 45s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f88f608d655049549d3b1fdbb340e648 : SUCCESS in 14m 14s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b67d6afc325b4a3aa1f49c56efec86b1 : SUCCESS in 11m 07s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/511ff25b84e14da2af37a4bde97b284f : SUCCESS in 5m 23s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4ec529efed0a4e0c86aaee7e3e97d2fa : SUCCESS in 25m 06s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2420178b8c6844dd8e8b75c8dce94113 : SUCCESS in 14m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7222f6db28a44ffbb7e913b1ab0a7b67 : SUCCESS in 4m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e846028464484b5cb6090aff99502fb3 : SUCCESS in 20m 10s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5825a3ba75e246099bf064c273567c32 : SUCCESS in 5m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2c8c0cd6cef540568237a3297fc7e1dc : SUCCESS in 6m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b362e99dddd5403b8e3d0cac25589c61 : SUCCESS in 11m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f34768fc47a04d8cbcd0412ec3501223 : SUCCESS in 50m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/812a2839c021485f8961f12c3a7f0de5 : SUCCESS in 1h 16m 08s","accounts_in_message":[],"_revision_number":3},{"id":"da593e1fdc5e4a09515fbd3e7f67d821df7fa2ec","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-15 23:01:15.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"eef8b8187d7783d217ffe742c0e09bf6a5a7e28a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 23:01:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"55877a7ba519dbdca8b845d2dbf5b3774d91bf5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 01:53:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/def63a3ba2764857a2f3fc5d24d0caf6 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a1f58fab176408184162b6e7a12fed4 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/920a56f573794140ba10bf0f66d484c8 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb10cf1f3c784b9d82777dfa11f965d4 : SUCCESS in 6m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4f69f1fd1fbd420a8517676c2687a30c : SUCCESS in 4m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fcee0fb5c6e14d0a89e88e308a1a23be : SUCCESS in 6m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d754deb49b544fd08babb19890de9b25 : SUCCESS in 11m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7b8b1d784e4640bca5fcd1ebedb6d125 : SUCCESS in 12m 25s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/757eadcbc859485e9d0bbccb21d3803d : SUCCESS in 10m 07s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d45ca49c71174ec78f4d102b700f007b : SUCCESS in 6m 01s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/92f6a0f378164cf88b75188d4a421692 : SUCCESS in 10m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bf009b39d7534d1f8ed5656f47935aa9 : SUCCESS in 9m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d08c9fd8023e46e49ddb0aea5ea62f8d : SUCCESS in 10m 22s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a2a253be2cbb4e0cbf3c1bff8225f668 : SUCCESS in 17m 07s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/81ef9c0f332e45cf91a23be4d59204c3 : SUCCESS in 9m 26s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d963bd9f9d8c4de083604b95df030e43 : SUCCESS in 6m 47s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4531ee6d4b4c4890b01b310e51185fe0 : SUCCESS in 26m 33s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f201d3dedcdc49c5a50c776dfae7ef06 : SUCCESS in 12m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f5f97ce2b9f144cd922c942c9baeec3c : SUCCESS in 5m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/44ef6a10c2fe47c19bd7760a10cc1ca1 : SUCCESS in 20m 41s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4abd6cc2baeb4dc393cee5850fd375ad : SUCCESS in 7m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/18b1927a611a4ecc9632b0b70552ab92 : SUCCESS in 7m 29s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1f2c2c7c9e2747d9ae20ab2fdab457f9 : SUCCESS in 12m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59aaa73d601a4062b4e60ef8122c3a82 : SUCCESS in 48m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf0ad7f11f9341fdb7f5c7480e4e4f66 : SUCCESS in 1h 12m 14s","accounts_in_message":[],"_revision_number":3},{"id":"51aa32a47802588565525d6b8c523613a90b55c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 01:53:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"5da64182b2efa59daa1916fbd8115f6ce9edf09d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 01:55:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0f61e35b7904cbd86707b2b7c35656e : SUCCESS in 48s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"40a9f989aceef50da91e5a10ca3c4352095111a4","revisions":{"1f124e59b4e978d4d51298172fa8052240b75921":{"kind":"REWORK","_number":1,"created":"2021-06-11 06:24:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/26/795926/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/26/795926/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/795926/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/795926/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/795926/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/26/795926/1"}}},"commit":{"parents":[{"commit":"22eabc5c2752e991f54316aba7ccb8dc09bd2cad","subject":"Merge \"update constraint for oslo.limit to new release 1.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/22eabc5c2752e991f54316aba7ccb8dc09bd2cad"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-11 06:24:40.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-11 06:24:40.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I2a240e6284b802bc830418ffafd8fdaa4a4a5780\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f124e59b4e978d4d51298172fa8052240b75921"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f124e59b4e978d4d51298172fa8052240b75921"}]},"branch":"refs/heads/master"},"8e7d068ffad09318ca5f8891674d0f62eb77a695":{"kind":"REWORK","_number":2,"created":"2021-06-12 06:24:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/26/795926/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/26/795926/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/795926/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/795926/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/795926/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/26/795926/2"}}},"commit":{"parents":[{"commit":"22eabc5c2752e991f54316aba7ccb8dc09bd2cad","subject":"Merge \"update constraint for oslo.limit to new release 1.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/22eabc5c2752e991f54316aba7ccb8dc09bd2cad"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-12 06:24:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-12 06:24:03.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I2a240e6284b802bc830418ffafd8fdaa4a4a5780\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8e7d068ffad09318ca5f8891674d0f62eb77a695"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8e7d068ffad09318ca5f8891674d0f62eb77a695"}]},"branch":"refs/heads/master"},"40a9f989aceef50da91e5a10ca3c4352095111a4":{"kind":"REWORK","_number":3,"created":"2021-06-15 19:35:50.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/26/795926/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/26/795926/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/795926/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/795926/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/795926/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/26/795926/3"}}},"commit":{"parents":[{"commit":"22eabc5c2752e991f54316aba7ccb8dc09bd2cad","subject":"Merge \"update constraint for oslo.limit to new release 1.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/22eabc5c2752e991f54316aba7ccb8dc09bd2cad"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-12 06:24:03.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-06-15 19:35:41.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I2a240e6284b802bc830418ffafd8fdaa4a4a5780\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/40a9f989aceef50da91e5a10ca3c4352095111a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/40a9f989aceef50da91e5a10ca3c4352095111a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
