)]}'
{"id":"openstack%2Frequirements~796252","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I69d378c04fe55264133b3cc3eb7bd31ddb379ea9","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"I69d378c04fe55264133b3cc3eb7bd31ddb379ea9","subject":"update constraint for tripleo-common to new release 12.4.5","status":"MERGED","created":"2021-06-14 09:53:53.000000000","updated":"2021-06-14 16:33:30.000000000","submitted":"2021-06-14 16:33:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796252-new-release","meta_rev_id":"1c2ff6cc5dc7ea3f1177ea635e258bd8a7b13bf3","_number":796252,"virtual_id_number":796252,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-14 16:33:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-14 14:46:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-06-14 14:46:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-14 11:41:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-14 14:46:00.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"8d1ca72967a3ca45bec25f525f13fc244fc0c9bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-06-14 09:53:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"012b460cbb40845ebc29397e7dacb0aac4646e31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 11:41:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/718dc901455c48e4ac1af43bea9a1e96 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2464c7a48ff4b3e8264ea11e2f48062 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e96f3b9cd0814fb08dc1ed31506b505a : SUCCESS in 5m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/69c8e50a3fe647bca01d4f9331a7f5a2 : SUCCESS in 4m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/55c3b54c3f14462cb8cc35915ee8f53c : SUCCESS in 4m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/13105e48c3a449be9633f0efd7b6c5ae : SUCCESS in 11m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a9b19fd6105c4d7eb1b94e74d2b357f4 : SUCCESS in 9m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/edefb34de4024658b6ee970a698a12b1 : SUCCESS in 5m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b3db7b9cfda1427194128ea9cd50cbf3 : SUCCESS in 13m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7e8c4f839a9844c189eeee96b17735e8 : SUCCESS in 8m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/909b74cfdba04192bbe4b618abe94e86 : SUCCESS in 15m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c9c8d46771014bd18ff66f0de7be49b2 : SUCCESS in 30m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f1ecf2fac02644f7993c7dd3144cd0fc : SUCCESS in 16m 17s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c9736fed832e47a789db462c8999fe63 : SUCCESS in 15m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/44ba7da86ed14d3a996b1fb1d0e91978 : SUCCESS in 19m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cfd075b027dc4d3b949d0c263da475b8 : SUCCESS in 5m 55s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b77be6dab1454982ac42e1c0148dff3c : SUCCESS in 5m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a67570dbafc64ab38d255c18b0357ba2 : SUCCESS in 16m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f1092864d56c42dba8c1ca046acf3330 : SUCCESS in 22m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1e5028d3693543b49fdef547d445032c : SUCCESS in 9m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/053a6c52916e4c62a0d0aa2f152e378c : SUCCESS in 56m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d3faf210a7434c5bb3d18824ceb773d3 : SUCCESS in 1h 06m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e8f057391ad4150a0ebd4588ab53a2d : SUCCESS in 1h 41m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/8cdc6e387b414d3ea4272e3bd848e7e8 : SUCCESS in 1h 11m 24s","accounts_in_message":[],"_revision_number":1},{"id":"003cdfc981dfffcc52f5a40c5313978a66a6cc03","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-14 14:46:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0fbb31683c919d429d629c4a20d00fba8167cce3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 14:46:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"44c6f3a451042df8831db68c145a61900e0cdcae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 16:33:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80abd73861ea446fb7119d78748bcf52 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4e27b368a5447a9b3ebbfc5f69e0801 : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9e390ac4e4ad4f5287fbef8d628c0015 : SUCCESS in 4m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ad2384ef69fd4a229fee7baac7e0cca0 : SUCCESS in 4m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/712d5e1d53f7431db9b99acd82ba0161 : SUCCESS in 4m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e72ea5d633f24491b2e2e50e114161a9 : SUCCESS in 13m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7f19f4acee1c4ebaa0ad626eb8c951e5 : SUCCESS in 9m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a4163f2c954a4bdeb8e7a3282c84d313 : SUCCESS in 6m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c98408bafa4e49aca118080895730cef : SUCCESS in 10m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6fce6159fdab446098bfc91734beb26f : SUCCESS in 8m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6e10b551bb6644e2bc28810710fb7bed : SUCCESS in 12m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/954d6d466d6b49b2a49a321a350ba890 : SUCCESS in 27m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dc7f59fa7d04420baf24bc244db0bdb9 : SUCCESS in 16m 39s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ec031f21aa5241cfa1c219782360f191 : SUCCESS in 13m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/43acf7fe6b1b45ffa2df0388cccd1fce : SUCCESS in 19m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9a0732ea3c1d41a7aa570e35720ac893 : SUCCESS in 6m 03s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/481c7c10193242b58c2aa6896be93e32 : SUCCESS in 6m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ed19c74c3faf4c58ada95358d39bb2ef : SUCCESS in 8m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e2b7d6be0d484f64b5dcb27b99e89f67 : SUCCESS in 12m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4d6b499996474867aa6c033d1ee4348b : SUCCESS in 12m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e75281a92bfa49afbbbb62dd993f8cf9 : SUCCESS in 50m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/50b6221c42314de4b52840f575271b5c : SUCCESS in 1h 09m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/790aa21ae2bc4109824498ab49c3bf51 : SUCCESS in 1h 41m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/792851d9296f457ebb21e3f00b8616b8 : SUCCESS in 1h 29m 32s","accounts_in_message":[],"_revision_number":1},{"id":"1c2ff6cc5dc7ea3f1177ea635e258bd8a7b13bf3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 16:33:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d66aaa8fc8891b65fe7840e4866e00ac7ae88133","revisions":{"d66aaa8fc8891b65fe7840e4866e00ac7ae88133":{"kind":"REWORK","_number":1,"created":"2021-06-14 09:53:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/796252/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/796252/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/796252/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/796252/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/796252/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/796252/1"}}},"commit":{"parents":[{"commit":"485a764df887e74e348674da1aeb461ab6feaaf3","subject":"update constraint for neutron to new release 16.3.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/485a764df887e74e348674da1aeb461ab6feaaf3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-14 09:53:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-14 09:53:52.000000000","tz":0},"subject":"update constraint for tripleo-common to new release 12.4.5","message":"update constraint for tripleo-common to new release 12.4.5\n\nmeta: version: 12.4.5\nmeta: diff-start: -\nmeta: series: ussuri\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: If0eb0b199288555fc58760f74a7afaa4da3d3828\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I69d378c04fe55264133b3cc3eb7bd31ddb379ea9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d66aaa8fc8891b65fe7840e4866e00ac7ae88133"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d66aaa8fc8891b65fe7840e4866e00ac7ae88133"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
