)]}'
{"id":"openstack%2Frequirements~796290","triplet_id":"openstack%2Frequirements~master~I89bd4a75336cee8d00e4041c40ce9f28cca16b9e","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I89bd4a75336cee8d00e4041c40ce9f28cca16b9e","subject":"update constraint for os-net-config to new release 15.0.0","status":"MERGED","created":"2021-06-14 15:24:27.000000000","updated":"2021-06-14 18:28:46.000000000","submitted":"2021-06-14 18:27:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796290-new-release","meta_rev_id":"5eb0b133b395696afb70dd1738b9e8c6d9567644","_number":796290,"virtual_id_number":796290,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-14 18:27:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-14 16:58:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-06-14 16:58:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-14 15:43:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-14 16:51:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-14 16:58:34.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"51bdb9c6882c4d10d81baa76502a96ab586a03d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-06-14 15:24:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a210c2b8ae5f012d78a04e197480e779458fa4d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 15:43:22.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5ebdfe72d7514dd2ab95d5daafa65105 : FAILURE in 13m 33s","accounts_in_message":[],"_revision_number":1},{"id":"25bccc29d8a07a66e27229bf425006923cddae2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 16:51:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aa09988a3b646578bdb83ea255514d3 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/054881735d1a47aea67d1999f3c84c10 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82c54ad9c0f1479f8ac0f269152c04a7 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0090844b7dbe4d2f864db17445f44a6d : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a2daed280e44847a89a0f285541c7c8 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d7aa4cb03e0047059dda58e4c1487fd9 : SUCCESS in 5m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/af8009bcc81547c8827f71e164dfa78a : SUCCESS in 5m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/85007a95bf774110a591d2accd34d3b7 : SUCCESS in 12m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e9cdd2f9dfa64695b5cb08ee0588fadf : SUCCESS in 12m 09s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a87ad809f4b943e0b934912449aef460 : SUCCESS in 12m 09s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/35fc1422c6b049d183848d49fd6845e1 : SUCCESS in 7m 25s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/2e9f73a0d4644fb3a2d9f19cfc93a4fd : SUCCESS in 12m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/66bb8d12d71c4541b9255f6e3281f56f : SUCCESS in 12m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/07e277b2ee4f497ba8486953c5e558f4 : SUCCESS in 11m 45s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5d0ad1f8ffd944e89045bfdc0295f07e : SUCCESS in 16m 56s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b4de5a355405440baf89548577ae7a40 : SUCCESS in 11m 06s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/4fb7c4e5e9ba4f49810bc7c7df74f2d9 : SUCCESS in 6m 40s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/66ab3e8ee33a403ca377523623e86d3e : SUCCESS in 26m 04s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ca150e6d023e4f4bbbc295d5bc5d49ad : SUCCESS in 13m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3cdb95f675204045b501589958028b66 : SUCCESS in 4m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a8e8ec79fed4451e95db1b3313d23021 : SUCCESS in 21m 02s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a3c793944e0d4c0aa9e47519084f31d1 : SUCCESS in 6m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5e75aad6a54c4620b91c5ad121ad375d : SUCCESS in 10m 15s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8e484c4b9d7843e59740d2350333fd12 : SUCCESS in 13m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fea468b806784b8ab4ab28effb741644 : SUCCESS in 53m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/42a0279cd06b4b7e9ae559172ce82e32 : SUCCESS in 1h 21m 20s","accounts_in_message":[],"_revision_number":1},{"id":"3769278d9ac33cb2dc23392d77d9ba998dddbd96","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-14 16:58:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"925c848c19699d4f88945d618f095123484d696e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 16:59:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"03fd3e8135f0379b900f18bfd817059c41d88f11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 18:27:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bc623ff00d74435ae313e890ed750f5 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19896b83321d41bc9c06f9e9c01e243b : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba015a6323e2429f8e9c0ebff6b9f15d : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96f3fc6b04f6453cb634b210d9c3838d : SUCCESS in 4m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1d1e3842706446da8ff5d32477c97c4d : SUCCESS in 5m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/105512de650b4d9b98b5798fbe5030b8 : SUCCESS in 5m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/72bd3f8e03c04e7e8bb9ae3dbe2d33d9 : SUCCESS in 15m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1558bef3acdb4f15ad31ae0d8f059076 : SUCCESS in 12m 15s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/590aef5abee5426fb5942d19cb690798 : SUCCESS in 10m 18s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/bf741243f2d04594b71aa61dba87b6ee : SUCCESS in 6m 27s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/cd5ae3c9492e4956805ec85ebd09f61a : SUCCESS in 10m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cf4dcadb0f7e40e5a96065c38bad83d3 : SUCCESS in 35m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4cc20068340f4eee9a655b3cc4084e1e : SUCCESS in 13m 48s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1c1c773584aa42adbb20d3455758e638 : SUCCESS in 14m 09s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9c0a6bb150a042b9aed56028a09a8891 : SUCCESS in 9m 16s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/40f05554ae1d45b59fa09c3472bd14d6 : SUCCESS in 5m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9bc44ccab2d44127815c0c7c576ed15c : SUCCESS in 23m 07s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d41b62f9c452488a9f3262b793710313 : SUCCESS in 13m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c24e1be4b0dd4478ac517ac728fec13f : SUCCESS in 4m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b4635c61605f40c39d311129cabe0e12 : SUCCESS in 24m 57s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1f1d6a5f434645269b4e335dc7ec7c29 : SUCCESS in 6m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8ffeb8adf3d346cdaac4a76be8bf2119 : SUCCESS in 7m 31s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9326c4ac3aad47b2a787dcb320ff3e0c : SUCCESS in 12m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da93ccf7016644aa814ab337fd544084 : SUCCESS in 57m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f1000a3023ab45e385200752d99a1e88 : SUCCESS in 1h 24m 36s","accounts_in_message":[],"_revision_number":1},{"id":"d3c27ff05fdbd2887dbc210493b9c090a8062179","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 18:27:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5eb0b133b395696afb70dd1738b9e8c6d9567644","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 18:28:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ad323ec057c48ed827228176893201c : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"40a247dc1f3e9c731552653c11867ca3d8a29278","revisions":{"40a247dc1f3e9c731552653c11867ca3d8a29278":{"kind":"REWORK","_number":1,"created":"2021-06-14 15:24:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/796290/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/796290/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/796290/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/796290/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/796290/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/796290/1"}}},"commit":{"parents":[{"commit":"22eabc5c2752e991f54316aba7ccb8dc09bd2cad","subject":"Merge \"update constraint for oslo.limit to new release 1.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/22eabc5c2752e991f54316aba7ccb8dc09bd2cad"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-14 15:24:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-14 15:24:24.000000000","tz":0},"subject":"update constraint for os-net-config to new release 15.0.0","message":"update constraint for os-net-config to new release 15.0.0\n\nmeta: version: 15.0.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: If80932ebd102d4bcda70256f4cf82cb5267939da\nmeta: release:Code-Review+1: Harald Jensås \u003chjensas@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I89bd4a75336cee8d00e4041c40ce9f28cca16b9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/40a247dc1f3e9c731552653c11867ca3d8a29278"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/40a247dc1f3e9c731552653c11867ca3d8a29278"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
