)]}'
{"id":"openstack%2Frequirements~796299","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I3f3052159d1498be27f22e1b5db1ccd69a161c64","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"I3f3052159d1498be27f22e1b5db1ccd69a161c64","subject":"update constraint for sushy to new release 3.7.2","status":"MERGED","created":"2021-06-14 16:45:15.000000000","updated":"2021-06-15 01:08:53.000000000","submitted":"2021-06-15 01:08:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796299-new-release","meta_rev_id":"f6c65d8837f4e54b0aed1a2ba69a85891b88e543","_number":796299,"virtual_id_number":796299,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-15 01:08:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-14 19:48:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-06-14 19:48:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-14 17:02:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-14 18:38:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-14 19:48:28.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"de4ade90642bbbacd6aeee8a202311741ebe9ad5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-06-14 16:45:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f733d1327e859cc20ccb7bf12cf408f7396fc2ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 17:02:52.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/202e6db2acca4395a8742f0abb198cc1 : FAILURE in 12m 29s","accounts_in_message":[],"_revision_number":1},{"id":"cc002528ccb7b7a0b78977b1452eea01bd1af7b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 18:38:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/392357b1767f4b72ae96fe0271348d19 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb366f923dc6492abde0d6617b6466ac : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/378396becd2748eb9c19c4ef8fcc5288 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73e01b38ea06427db04eb812f5353588 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/621b922fdba7467c9324d6836dd5cef6 : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8030ae92c36c4c17b639465958ac097e : SUCCESS in 4m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0d6a7d8d415a4b88b1187e7f896779f5 : SUCCESS in 11m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/99053b6a79db40d5976502b4e9bd5e70 : SUCCESS in 9m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ea7476565e764e88af1b1ac5f25c857a : SUCCESS in 10m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e5c090b76ff14f14b85c44ba70a7fbbd : SUCCESS in 6m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6076d97770ce412597dafe8cb39d8b20 : SUCCESS in 9m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5fc9ba637454475a9d7836d78e6d9c82 : SUCCESS in 8m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/07c5f69e457842b2b533af89cbb04003 : SUCCESS in 9m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a4bfa55018044193bbbe7080a68bca3b : SUCCESS in 12m 21s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/7f104a0e735440249ffbbef55614857d : SUCCESS in 5m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/832d21184c864e00bc6dee9f7de9c9d2 : SUCCESS in 24m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d5daf65d92f54c05ba5ac051e95f1a3d : SUCCESS in 10m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5c413929e9ca40818e086117d0c97cf5 : SUCCESS in 4m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/128aa80a2de34032bda79d938a47ccb3 : SUCCESS in 17m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/75669766c96c4ae0b52506edacc88c03 : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/57dc352f35ac4381b64871807545489b : SUCCESS in 6m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3a817e01986c48a9aa000e5e110f13bf : SUCCESS in 10m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a509d7e5e7d94deba555676533cdf6ef : SUCCESS in 55m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d94d081068a4b80a1a25e6fef3192eb : SUCCESS in 1h 49m 13s","accounts_in_message":[],"_revision_number":1},{"id":"852f9375673b864538dce5f988f77cf592311b90","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-14 19:48:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ac607479085ed6e8c7523dd1b0da5fec5e9b272d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 19:48:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ecf34d781877b3bedc599ed71e36ecefe88db56b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 21:14:24.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae2e4e83c2584fa4a4157357d2f9228d : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d129f1332ae4c26bf45bd91fda10102 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bce72a4ec9f24c11bff91345768d97c8 : SUCCESS in 4m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6a79b4854f6c40b0962517f8324a3ea7 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/148c733d4ba9437ab6bd726dab23e87b : SUCCESS in 4m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d2ed423676274e44836cef85094a1677 : SUCCESS in 9m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0b9988bca57a417bbbde2d4a3bf68087 : SUCCESS in 9m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/002e8b8f1f5c41869cf3e396ec88b38b : SUCCESS in 10m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/534da607ff4446ce816689c92d683590 : SUCCESS in 6m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/74e47a615548442c8fbf9f37a7ec4d0d : SUCCESS in 11m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/675ca6548e8c4ec3ac65f6dd74b81af6 : SUCCESS in 8m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4ffa0e88cf9447cc8a6af14f45e37ea5 : SUCCESS in 10m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a1101a2480b147cda24fe55b1efd83c9 : SUCCESS in 13m 18s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/a7489da85ccb4ecf9794ca070fc75ca5 : SUCCESS in 5m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/73d4d812f04a4b2e916349bdf0c9781f : SUCCESS in 31m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9338b64b2f444448b66688f729cd3694 : SUCCESS in 13m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7c5c1f6c9bf348c2b8779e794b6b0a71 : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fc4ac4cd0f8e4fe5a6eccc9807753227 : SUCCESS in 18m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/714ed2aba4ed4803b3740e331c42d40e : SUCCESS in 5m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/727d09117ac94bb68c9a376d0d6ef143 : SUCCESS in 6m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8c547fea03704ebdb3a6017ba03404ef : SUCCESS in 11m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf4eb3298ece49f7893bc58e6d740dfb : SUCCESS in 52m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a22bcb867a75437faed04644e0ea3316 : FAILURE in 1h 11m 19s","accounts_in_message":[],"_revision_number":1},{"id":"89ef57fc72e4d6044e1bf3a5ef44e0df49e6830e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-14 21:48:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"859aa6de0f34ac5c7e84db434d2c5e1019fa8a35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 22:02:10.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9dcee0d598fb46bc8f2dc8eb0cf2e78a : FAILURE in 9m 40s","accounts_in_message":[],"_revision_number":1},{"id":"4b91d5cdc3ced0d07ac6b85603b0bb6897881eb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 23:18:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28a979f01a1841e494c5c9d742f96398 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f8c1abe1d54468e9efba359f643767b : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff4e93374af84d3d9c7b2889f9c315b1 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74af9d81e6f74b15b5bb34296eafe5ae : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b17b364c5995440396a4e068021c76fe : SUCCESS in 4m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f2a1768d314f456cbabc024f128b2eb9 : SUCCESS in 3m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f7c8f999e9cf46eabdfd3938213d8efb : SUCCESS in 9m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2822897d611f4ca3afc60b618badc349 : SUCCESS in 9m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/eaa80de64aa4413bbb74f0feb05a7818 : SUCCESS in 11m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e640af271e4a4f4d9a9799d3a1d83e97 : SUCCESS in 5m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/265fb30519db48f3812f7839b9856cd0 : SUCCESS in 11m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/91dc83e3c600407080843107a623c0d1 : SUCCESS in 7m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7372bde792504593ba3f8cf70ddcf3a9 : SUCCESS in 10m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3cd8b0ac45bd492abb498f0778f71df3 : SUCCESS in 14m 57s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/9e8c486a492647b782b9c60e94f24edb : SUCCESS in 6m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/aa061d35e7064aa2be3e47b1b27c0b3f : SUCCESS in 28m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/eb3b52fc53e24ad189584bc4c48ce19e : SUCCESS in 12m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b90022e9406d4f9aa6160e2196fc458a : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/432b8b7b5a0541ba8c632483f75596fe : SUCCESS in 16m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0587e509ad4d451ab0976717650fe847 : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c9a8373ae35e4c999cbb915fbdf20b7e : SUCCESS in 6m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/28a6a65a12a34f9484f5e3981cbdf31f : SUCCESS in 11m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/012f36d734e449efbd911857bfba0a43 : SUCCESS in 46m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5f4ccd2c58eb43ae9f2bb438133ba5e6 : SUCCESS in 1h 28m 00s","accounts_in_message":[],"_revision_number":1},{"id":"b745f8d7d4dfd9181e7adb8e3f162d3c48345645","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 23:18:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"033b94fba33aaf80bd3f03cea5191200b21c0c7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 01:08:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ac783da871e488c9ed30559bdd105d5 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fb8f73d45984bd0a7b41ccfa410b9b8 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1512ee0b0e04170ab6843b960442fc4 : SUCCESS in 4m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/88b6d2ce799440cdb3cbacda5fca4301 : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/94718bddf3ae41a7984248cdc0163b3d : SUCCESS in 4m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3d57cd19aa02459fb6f64198a67f22e6 : SUCCESS in 8m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9986406184464cf98a1ae78df7475add : SUCCESS in 9m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/53c59afe507b4040be7c898f64224146 : SUCCESS in 9m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c387f6a2d380437f8741aac4e4c4cdc5 : SUCCESS in 5m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/58b9a5c7ef8b42ba839eeaf862bbc685 : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/af6ccaa595ac4d0ca525e28510bc32bf : SUCCESS in 7m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b005b37a2adb48378453efdb182cea60 : SUCCESS in 9m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cdfe7f391b604dd98fcfae91183c028a : SUCCESS in 13m 12s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/0785787bbada47919b435dd771839211 : SUCCESS in 5m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/28bade79a4014d399e36ea6bcc2ab126 : SUCCESS in 25m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/829f22e492bc432ca8948bff3924cde7 : SUCCESS in 10m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/aa61ab2ba2954bfcacf1a0113f577c2f : SUCCESS in 4m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5ec63a833c1147cb95bc5c6fdd256456 : SUCCESS in 15m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c9da74d355404854b9c20f24a8a23c34 : SUCCESS in 5m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0302fea19f204443ba2a4952a0869d64 : SUCCESS in 6m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1b9790bbca984368bd7865c47dbb0575 : SUCCESS in 10m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3cdd117c51af4af897172a2d7183a21b : SUCCESS in 43m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/675159731add487b8aceb3097a558174 : SUCCESS in 1h 19m 57s","accounts_in_message":[],"_revision_number":1},{"id":"f6c65d8837f4e54b0aed1a2ba69a85891b88e543","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 01:08:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aee7a4b34ee91065581ae20ba81d30920bb18769","revisions":{"aee7a4b34ee91065581ae20ba81d30920bb18769":{"kind":"REWORK","_number":1,"created":"2021-06-14 16:45:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/796299/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/796299/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/796299/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/796299/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/796299/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/796299/1"}}},"commit":{"parents":[{"commit":"c4ef65769ebec9a8d32915cc0dcf5951c3eb4cf4","subject":"update constraint for tripleo-common to new release 15.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c4ef65769ebec9a8d32915cc0dcf5951c3eb4cf4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-14 16:45:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-14 16:45:11.000000000","tz":0},"subject":"update constraint for sushy to new release 3.7.2","message":"update constraint for sushy to new release 3.7.2\n\nmeta: version: 3.7.2\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Commit: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Change-Id: Ifd8fcc669a7c2761b1b6f33a8176241cb2a59c3d\nmeta: release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I3f3052159d1498be27f22e1b5db1ccd69a161c64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aee7a4b34ee91065581ae20ba81d30920bb18769"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aee7a4b34ee91065581ae20ba81d30920bb18769"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
