)]}'
{"id":"openstack%2Frequirements~796300","triplet_id":"openstack%2Frequirements~master~Ib910c14a0921b5d5fce8079de070c594540ecfea","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib910c14a0921b5d5fce8079de070c594540ecfea","subject":"update constraint for sushy to new release 3.9.1","status":"MERGED","created":"2021-06-14 16:47:20.000000000","updated":"2021-06-14 19:37:42.000000000","submitted":"2021-06-14 19:36:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796300-new-release","meta_rev_id":"5fa28c715e01590ef5f391753826f81f4f8feeda","_number":796300,"virtual_id_number":796300,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-14 19:36:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-14 18:20:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-06-14 18:20:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-14 17:05:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-14 17:59:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-14 18:20:20.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ab00baae96f107d4f238b48e1093857b58137480","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-06-14 16:47:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"433e296af2fdc035bc3a27c9fe2d59c83ae0fe3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 17:05:29.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2d8acabbc2bb4355a1ccaefe20cc2916 : FAILURE in 14m 06s","accounts_in_message":[],"_revision_number":1},{"id":"6e7dc0fca2c34c0036433a5aaf67466b6306db80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 17:59:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/069ac1bc0a19430598290ddbf69cb906 : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d72ad1d389e4c5886f7c19536b7004a : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47c49a84f18e47ac961cd85cc1f38761 : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57c236329a404a18a3fdbb607ba8472f : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a9dafaf5a6d49f8bee6ab92334a905f : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c5d637315cde49ed80749d6b182ad74a : SUCCESS in 5m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7c06f3f85e3842b0a657294d84e8342b : SUCCESS in 4m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c211a0cba0aa4554b7b8bc667f649b2f : SUCCESS in 13m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f146d53f3d0e4efaad6e73f31829646b : SUCCESS in 16m 15s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b117bd1b0a3c4de2a4864d21ae0aeba0 : SUCCESS in 17m 19s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/07e7e14aec7c48baadc998cc2e7f08cf : SUCCESS in 5m 35s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ed10f666a67a4a4eae2c1bf79725d6c3 : SUCCESS in 9m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a8b0794c2eb446ea8067f032977378e1 : SUCCESS in 15m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/dd34e48aed324b8ca9bf9ab58e8bc1a4 : SUCCESS in 9m 24s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1c23b1520967469e8c496781e2d7151c : SUCCESS in 17m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ce0337acbe9c49bdb4698c8ceb39807d : SUCCESS in 11m 56s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f420118de5014c618bae3b2bf480f3ca : SUCCESS in 5m 16s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/894c6cac1de04f43973e0b7a876a9619 : SUCCESS in 30m 57s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/48ffa5b0bd9a4bb8b7e1961dba03e1b1 : SUCCESS in 15m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3c34e3cf6bc242c6a72d4397d9fe6a53 : SUCCESS in 4m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/16f5d680aca743ab9f7802abb2e720a0 : SUCCESS in 24m 21s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/bc5a09244e3e472d9b0ff0e827a8b7be : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/433f9031d09a478ba0ab98852e7be089 : SUCCESS in 7m 11s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0ec9e148249c49a3b049d1618d7e9a92 : SUCCESS in 10m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9dd9fc0c5adc4e948467fdd05348f9cb : SUCCESS in 45m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e4c2e0bd55b4d4fa9b66fbc13aea351 : SUCCESS in 1h 11m 17s","accounts_in_message":[],"_revision_number":1},{"id":"2105c24b0892e8d98718945102f2153775a64278","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-14 18:20:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a71739461a935fcb1b06f66bd9684f18ff25bb85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 18:20:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"401bee645882492d744f41ad08f04c1a600f716c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 19:36:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e83f30e25684a16a301667d460cb7d0 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf8dc4bd37e040fe9f39698d1bdb6a58 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f179bf30023c4fd38664f94845a09c20 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2ecec129f39457692af54260f79a1c7 : SUCCESS in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a122c925776d4732b3f7709160a9c7e7 : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/711d084aa6b04c6491ddb64d839fc16c : SUCCESS in 4m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a5ce62cd79e04adeb739e4dd250d7dc1 : SUCCESS in 12m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0615944eb69640e489ed4b494ecc7ad8 : SUCCESS in 10m 20s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/cfcba3be2952479798f506532b0f7ce9 : SUCCESS in 11m 11s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/46a8319c4f284dcc816bccbeded4ba26 : SUCCESS in 8m 48s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ebfb0c3c82064dc2af0f8aa138ca991f : SUCCESS in 9m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ec69522f84b04dd48e7826bdc7023b5e : SUCCESS in 8m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ffc4282c82a545c98b38f3ae50ada4f0 : SUCCESS in 9m 01s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/9af5614907d84941a96ba183655e220a : SUCCESS in 14m 59s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/19af9321b25f46e7b22d3c68ca7ff8d2 : SUCCESS in 9m 01s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/3131c4c8bf3a487ab86ca5a5a8b89e87 : SUCCESS in 5m 42s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1e5ae2f71ce64d23afc980b3e329097b : SUCCESS in 29m 41s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2c3bf0d9772040839fd3a19851f45c29 : SUCCESS in 12m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/994ebcab422348f9a4d63025a77d0fa2 : SUCCESS in 5m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/80db82f5819b4886bb1c4f3c83a17222 : SUCCESS in 21m 39s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/fa02f244f15249fa82d06c73044e9a71 : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/534b2d206d7749b68ecbdd8ad711d2c0 : SUCCESS in 7m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8de3d20335be4c9db8549e189d7654ff : SUCCESS in 10m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b588afdcf0834a51b9c69eb8ac8406fa : SUCCESS in 46m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/975c046d7b4c4d4c8bd73ae037ee5f14 : SUCCESS in 1h 14m 55s","accounts_in_message":[],"_revision_number":1},{"id":"39457f5c725113e8b27804913cd9a798e13ecd2d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 19:36:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5fa28c715e01590ef5f391753826f81f4f8feeda","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 19:37:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0075d2b6eeaf425ea40330061f781935 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"316b9870e13caa1711ff7c56f0eedf4433072f4b","revisions":{"316b9870e13caa1711ff7c56f0eedf4433072f4b":{"kind":"REWORK","_number":1,"created":"2021-06-14 16:47:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/00/796300/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/796300/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/796300/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/796300/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/796300/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/796300/1"}}},"commit":{"parents":[{"commit":"22eabc5c2752e991f54316aba7ccb8dc09bd2cad","subject":"Merge \"update constraint for oslo.limit to new release 1.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/22eabc5c2752e991f54316aba7ccb8dc09bd2cad"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-14 16:47:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-14 16:47:18.000000000","tz":0},"subject":"update constraint for sushy to new release 3.9.1","message":"update constraint for sushy to new release 3.9.1\n\nmeta: version: 3.9.1\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Commit: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Change-Id: Ifd8fcc669a7c2761b1b6f33a8176241cb2a59c3d\nmeta: release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ib910c14a0921b5d5fce8079de070c594540ecfea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/316b9870e13caa1711ff7c56f0eedf4433072f4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/316b9870e13caa1711ff7c56f0eedf4433072f4b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
