)]}'
{"id":"openstack%2Frequirements~796301","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~I2d19d10aea7fb9354376829cbd78b71030a3e235","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"I2d19d10aea7fb9354376829cbd78b71030a3e235","subject":"update constraint for sushy to new release 3.4.4","status":"MERGED","created":"2021-06-14 16:50:37.000000000","updated":"2021-06-14 22:27:52.000000000","submitted":"2021-06-14 22:27:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796301-new-release","meta_rev_id":"c1a468c79c70c5340c7893edde3d960151674086","_number":796301,"virtual_id_number":796301,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-14 22:27:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-14 19:48:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-06-14 19:48:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-14 17:10:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-14 18:35:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-14 19:48:29.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"dbf654a60015bc1cf5a807aa21f01da6365de53c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-06-14 16:50:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1bac42ee11e4eb71000526c2c46c46cba29c6753","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 17:10:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a438709cac6c4b7d971518f02568df04 : SUCCESS in 14m 49s","accounts_in_message":[],"_revision_number":1},{"id":"d1dd0a921ffaddf159d35a5d3a709faeff53dd49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 18:35:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2043faa5be144af69004e010ab25520b : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8970a3011254049998713915a7b1922 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e3f4d6e2ae04df78757587c27569e1f : SUCCESS in 4m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3c24c8b2aa974e95ae99b02f5e7a8ed9 : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/83bf6691f8bb43d8bde688d62d43577b : SUCCESS in 3m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c8e9ddb411d047db97c098229415417a : SUCCESS in 8m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fd605e98cc06466a8c29fd6ea39acfe1 : SUCCESS in 7m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f15b570d72ff49ec8a8b7613f370143c : SUCCESS in 15m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3d854a4e20f042b8ad28b5b0bd055943 : SUCCESS in 4m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/31658f4b997641b1a08a7b620429882f : SUCCESS in 10m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/16dc155c55cf4c8aa3e738ad0c870b5a : SUCCESS in 9m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cbecf32f04f94f06bd86f816175e6fac : SUCCESS in 7m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e934038fbc754b22bf4778f3f5012dc6 : SUCCESS in 11m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5244d5e6264c441394055b00f08303cb : SUCCESS in 23m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bb363197f8d24b8dacb4f89df3d0ac01 : SUCCESS in 10m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7c62cfb1750840428c16476365303d2b : SUCCESS in 3m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/03cb84c748af437585a9f0149a8323a4 : SUCCESS in 18m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a38b20467dfd496fbd0996eccb064ffe : SUCCESS in 8m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2b705bdb6e584b249b9553723d667efd : SUCCESS in 6m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/85d45fdc36074eadb8774f4b06066fcb : SUCCESS in 14m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1864c981250d40ed8b962943fd009989 : SUCCESS in 1h 01m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4c8c51f3b01a4a44800f370c26baca0d : SUCCESS in 1h 38m 00s","accounts_in_message":[],"_revision_number":1},{"id":"e2f683dcbb7b59a1d61837eabb696cf911f695fa","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-14 19:48:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9a5df928e8053ef61ec158c7f5edec3b61f6564e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 19:48:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4fd749c7f5a9638c1de1e71b7fb3c8df162f9cd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 22:27:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0676f644137a4d6ebfa184d330802b11 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6eae72d62174d6ab8577ab33dd89605 : SUCCESS in 8m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/987d9b42621d44c79f679f97c9a37ff1 : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7c553ce2c9844cc6b300505077770e7f : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1bc6ea761c4a4e07ace578eb784fa17e : SUCCESS in 4m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f272aa90b1a848868478e3866f8d7206 : SUCCESS in 9m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5f177ca19a4547d8849142be01b756d8 : SUCCESS in 10m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ea50c3b6c6734bfc8c757ae5bdf5c5ed : SUCCESS in 9m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5d85860d40e645529e395a7b61206c1e : SUCCESS in 5m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/11dcb1a8d00c49c5a01c98634f2c8780 : SUCCESS in 11m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0bdf012dd8554de9a736c01ae8d9cf9f : SUCCESS in 8m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/82c6d2b0c8d041dda8e8769c132e9aec : SUCCESS in 8m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/55201b9570da4b048ff0c6c8d1974adc : SUCCESS in 14m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/be8c7cefc5ad4948b14e5416e9bbb54b : SUCCESS in 29m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2ca48cdd751e487cbdda92650617034f : SUCCESS in 14m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3c1720e44b684494b5910152f85c8db2 : SUCCESS in 4m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cec932f399bb4fc8b61ef2feedfaf9e6 : SUCCESS in 19m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cffffa10d3b24cf5ac121b44c781b39c : SUCCESS in 6m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/947a504d01c94ae09a9ed3895fa46c96 : SUCCESS in 7m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/28387cb2568740ba9f9d7c56f19842c2 : SUCCESS in 11m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c10f033489eb4cbdb49e26ef90e493a1 : SUCCESS in 50m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41fc8f1ba2bf48558092d4367eda794b : SUCCESS in 1h 23m 16s","accounts_in_message":[],"_revision_number":1},{"id":"c1a468c79c70c5340c7893edde3d960151674086","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 22:27:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f11be382b0e9608e2053308ada263db1e5506755","revisions":{"f11be382b0e9608e2053308ada263db1e5506755":{"kind":"REWORK","_number":1,"created":"2021-06-14 16:50:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/01/796301/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/796301/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/796301/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/796301/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/796301/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/796301/1"}}},"commit":{"parents":[{"commit":"9109b6be72339b5af0759cf455e3ce36477b0181","subject":"Merge \"update constraint for os-apply-config to new release 12.0.1\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9109b6be72339b5af0759cf455e3ce36477b0181"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-14 16:50:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-14 16:50:32.000000000","tz":0},"subject":"update constraint for sushy to new release 3.4.4","message":"update constraint for sushy to new release 3.4.4\n\nmeta: version: 3.4.4\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Commit: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Change-Id: Ifd8fcc669a7c2761b1b6f33a8176241cb2a59c3d\nmeta: release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I2d19d10aea7fb9354376829cbd78b71030a3e235\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f11be382b0e9608e2053308ada263db1e5506755"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f11be382b0e9608e2053308ada263db1e5506755"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
