)]}'
{"id":"openstack%2Frequirements~796311","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~Ib515e03312f27f208d4fbc8ebe9fcf77b3ebe352","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"Ib515e03312f27f208d4fbc8ebe9fcf77b3ebe352","subject":"update constraint for tooz to new release 2.7.2","status":"MERGED","created":"2021-06-14 18:36:40.000000000","updated":"2021-06-15 00:32:31.000000000","submitted":"2021-06-15 00:32:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796311-new-release","meta_rev_id":"51cf0fa71aaf9d16db1f5a63e00c8534947d34f2","_number":796311,"virtual_id_number":796311,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-15 00:32:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-14 21:48:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-06-14 19:26:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-06-14 21:48:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-14 18:57:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-14 19:26:26.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-06-14 19:56:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-14 20:06:30.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-06-14 21:48:19.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"0e08e40457a668515af3ed13cb0467df657ca725","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-06-14 18:36:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81314bcd3ef40e29ed7e531df9c9422cd0d0b556","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 18:57:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fee38a73cd794d06bb399d1bab220a7d : SUCCESS in 17m 19s","accounts_in_message":[],"_revision_number":1},{"id":"7ec0da38bfb12adc24d13eff80f47851bb02945a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-14 19:26:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6db831505aaa7a59d83287f0144fd159e3078381","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 19:56:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48fb0d6307584b19b863da434f293240 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b78ebcccf551488f839000aef2b10d47 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ea391829658433f97af0fb9c57c6341 : SUCCESS in 4m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e6935721b02443318405ca70089ba06a : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/be4d68e248b24004ae7afcc3c3619c2e : SUCCESS in 4m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5764c7606f2b4294b22b9e9032e16fdd : SUCCESS in 8m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/58eccf0ceaad409abe19938b88ec250c : FAILURE in 15m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c1d2f27cafbc4a9fb688edb733ab55f7 : SUCCESS in 12m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2336a19d2b4246bc96bcf1b2c834bd91 : SUCCESS in 9m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d2216bbae8da4076957201ce6ff7d461 : SUCCESS in 9m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bab375cc473648638982d08d60bffc95 : SUCCESS in 9m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7c247839177e4c65b5adefff9cdce633 : SUCCESS in 8m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f948e9a8c99d40c08f941e38b9bffa76 : SUCCESS in 14m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/725ca0c2d646407394980a039936c605 : SUCCESS in 31m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/db7927f62045406da22899b1b7985ed8 : SUCCESS in 13m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/41928a920319489cbc0e8e5658276e84 : SUCCESS in 3m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/439d86776cb64d9aab9002aecd513994 : SUCCESS in 17m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/63f1d11b8e4b44e287f75515b6e88534 : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2b363f4adc4b4f47afdf9a1dfab3efd7 : SUCCESS in 7m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5d89418d1cc04aeab8ca197707d6757d : SUCCESS in 12m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8f956fdeed44b608a654a9ffe5848da : SUCCESS in 56m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e5d5b8197acf4d5bab81bcf74e7b212f : FAILURE in 23m 31s","accounts_in_message":[],"_revision_number":1},{"id":"ecf68a5dc91acb85724c42c20366af8348be1b48","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-14 20:06:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f229f708cd26b3c810e222bb955f418190b23b19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 20:24:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/14c818c9de8e4b2a80f2cb9dd37c949a : SUCCESS in 14m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a9b9b9b38574353b6ef1c4d1039ef050816d16c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 21:46:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/432c57c5f5cd4cbdbc6c6fe76e28073d : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/663bb0d4e3d348cdb95d89d79bfaebcb : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf34f33983aa4106bac9e7ee9fd8dd28 : SUCCESS in 4m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d84bbae8eeb149baa94a079437505a5e : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b89d6e29c8114c3797342cf4b2edfe58 : SUCCESS in 4m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f6b16474f9e5427aba44e5d17670342f : SUCCESS in 9m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/603b5b4b1d6d4dbc86ab723f3842d03c : SUCCESS in 8m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c4716b72622846509ef19b5ce7030f67 : SUCCESS in 9m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f1b381c6206345f7a61b3868fa4127b0 : SUCCESS in 5m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1dd8ee3234824c9cae57d8ef4503b14e : SUCCESS in 11m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8e1e61b804e746009927bc09720c0791 : SUCCESS in 9m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0152a49106e041bd905bb5e0607499ff : SUCCESS in 7m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ccfcae2ccaee4141bef08d6aa346ac8e : SUCCESS in 14m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6539a3c63eed4e15ab56ee6f6c39d6bf : SUCCESS in 25m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/67628deba008436a9a961b3fa424b526 : SUCCESS in 13m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b38c6c804492485689cb030533f107cf : SUCCESS in 3m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/785df493d97541a6ae0b30a5407fa3f1 : SUCCESS in 17m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7367e7b3ca1a445cb84ce7968f3879a6 : SUCCESS in 6m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/850db9f6a9624e8abffd3c43810dcdaa : SUCCESS in 6m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6810db2f2dab4540816e64eed9b75032 : SUCCESS in 11m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0e2df104b16648e697e00219fca3b70e : SUCCESS in 54m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d16ca71304ff4f45ba242f0864b89816 : SUCCESS in 1h 36m 00s","accounts_in_message":[],"_revision_number":1},{"id":"5dcf4c2d08b1c87da87a5bae264b9719184c3838","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-14 21:48:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0f52434b1017a321167f639574d07836232e4f65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 21:48:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c5b6bf4bf6e61cc1794daee2a5b6851f003d8525","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 00:32:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6da10a9b99d942ee994c926bd69e8113 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f84b953bcd7416b910a528718fed153 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/827f77258fb345dc9f4da0995ba923de : SUCCESS in 5m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f03fef9dabbe4a5b8504eadac56c6367 : SUCCESS in 4m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6c8e548ff13241e5b0db7e367d9da0dd : SUCCESS in 4m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/12629e6dbf6442de8266753b1763c6b2 : SUCCESS in 11m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e80e8b7bd78b4adcbb5a4bc7f365f442 : SUCCESS in 8m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/829013935250441e9b2e3a026768e897 : SUCCESS in 11m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/397b2dadc464435e93760b1221875b5d : SUCCESS in 7m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/27576e51e5194e2ba69e777e94afb7de : SUCCESS in 13m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6f2baf01926a4f0ea343436f0bdafadb : SUCCESS in 8m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e7f72ce1a7c04a658c92b0e630e77f14 : SUCCESS in 9m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7da5c66ad9a545dc91eeb838f0111a80 : SUCCESS in 16m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6953e6544b8a47b4a57a509c2e85efd0 : SUCCESS in 22m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fc78180765d44ba9a16325ac21ad64ac : SUCCESS in 16m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a500f7b774e54f0db63417423738033d : SUCCESS in 4m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0fc0457542da4ac285b83027d5e20da7 : SUCCESS in 21m 07s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7cfa6c928988442c89ce89c9a6b8d65e : SUCCESS in 6m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/158ad41599f3424881ceb2d7c19c2815 : SUCCESS in 9m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b49a20de183e47eab06596a80757ec72 : SUCCESS in 12m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d6ee7a0757c84ee8b0da0aab09830ee1 : SUCCESS in 53m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/55b38e29630b484c938ebb9be219da0a : SUCCESS in 1h 23m 49s","accounts_in_message":[],"_revision_number":1},{"id":"51cf0fa71aaf9d16db1f5a63e00c8534947d34f2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 00:32:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cb29d8d3958d7d305a77c5bcf2062e593aaf591b","revisions":{"cb29d8d3958d7d305a77c5bcf2062e593aaf591b":{"kind":"REWORK","_number":1,"created":"2021-06-14 18:36:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/11/796311/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/796311/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/796311/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/796311/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/796311/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/796311/1"}}},"commit":{"parents":[{"commit":"9109b6be72339b5af0759cf455e3ce36477b0181","subject":"Merge \"update constraint for os-apply-config to new release 12.0.1\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9109b6be72339b5af0759cf455e3ce36477b0181"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-14 18:36:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-14 18:36:39.000000000","tz":0},"subject":"update constraint for tooz to new release 2.7.2","message":"update constraint for tooz to new release 2.7.2\n\nmeta: version: 2.7.2\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I588e7daa47b27e98adf28d4242384efd9cc58c07\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ib515e03312f27f208d4fbc8ebe9fcf77b3ebe352\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cb29d8d3958d7d305a77c5bcf2062e593aaf591b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cb29d8d3958d7d305a77c5bcf2062e593aaf591b"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
