)]}'
{"id":"openstack%2Frequirements~796416","triplet_id":"openstack%2Frequirements~master~Iba80669226464500778b9d2048252bde0a55145a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iba80669226464500778b9d2048252bde0a55145a","subject":"update constraint for oslo.db to new release 9.1.0","status":"MERGED","created":"2021-06-15 10:06:09.000000000","updated":"2021-06-16 03:30:38.000000000","submitted":"2021-06-16 03:29:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796416-new-release","meta_rev_id":"b702b88b947572e514e4e59709663063b5a98ab9","_number":796416,"virtual_id_number":796416,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-16 03:29:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-15 16:54:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-06-15 11:11:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-06-15 16:54:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-15 10:55:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-15 11:11:05.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-06-15 11:43:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-15 15:05:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-06-15 16:54:51.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"21e2ee2c633a366c106a7d67bacfeb56ba29efc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-06-15 10:06:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4264de5ea86c96c5c634a1bfd97382f07e54d2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 10:55:07.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b670dac2efee40d3997ee84acd3e2ce3 : FAILURE in 19m 49s","accounts_in_message":[],"_revision_number":1},{"id":"80b93d7e3dff919fd5e6facaceb92c249399e128","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-15 11:11:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2af7af87ef3a13de4541015f089cc1d1bc47f044","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 11:43:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/104f9f47d01246cd81b514b8374704f4 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c05f1fbe5df462d85745fce757fc467 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bcd8e7c76224ddd9e656f9c6e3b7ab6 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0f8ff2a56d847e0bb5b86308916692b : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6d11932e0b44fcbbdcefd92fe81e264 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d500a0622a0f40368882ec1201c42c06 : SUCCESS in 6m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d5849d244dea4f588b9822358e50bb4a : SUCCESS in 4m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d6b6591f46814e11930de3e04887c521 : FAILURE in 29m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c9b8eedd38e445e0bc18fcb4e076dcb8 : SUCCESS in 11m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e55aa38064854caa90cc9cb42ce34206 : SUCCESS in 10m 38s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/97b069359f3740b288989346dddc3b54 : SUCCESS in 5m 50s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/31ce2b9011fd4e139101d57729163bb6 : SUCCESS in 10m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ea4eaa391d24437ba7e6926679589c29 : SUCCESS in 8m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/714b4388076f4804854f5c0bd8b6933e : SUCCESS in 11m 25s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7354f7344ebb4451a93d2e939b8629f0 : SUCCESS in 18m 16s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/73c0a8ef747245f18ef6e3b7a7138975 : SUCCESS in 9m 41s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/79772951028c4c74aa24bbb38ba8c6f6 : SUCCESS in 5m 16s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/24fadd65353e461d92335a9f331d8d59 : SUCCESS in 33m 14s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a01c4ad67d994226bcfc3aa607a8e7f4 : SUCCESS in 13m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/38195889b29c4fbdb6bcd84aa46f9157 : SUCCESS in 5m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c7c75364ba1f4d69a011f9870348cbe1 : SUCCESS in 22m 21s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a351648b238841ea9020e89da364c264 : SUCCESS in 7m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6d835edb91624194b5baa0e7763adc50 : SUCCESS in 7m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7cb6f8b4478a493f84c8a7624d2bd050 : SUCCESS in 12m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/06751e690c4d498b8cd3cfdd70db651b : SUCCESS in 51m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3c03739a882e4d438b8576c52b591da0 : SUCCESS in 1h 31m 26s","accounts_in_message":[],"_revision_number":1},{"id":"71a8708876fd37b157578fa6025623efd5a46ebe","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-15 15:05:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6fd3b8c53cabf5ec3c94b5cec04f936baf8818e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 15:50:13.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/13c08a7916494804a31d3c3ae69494e4 : FAILURE in 12m 42s","accounts_in_message":[],"_revision_number":1},{"id":"505a9191f8c3ccb674e3f97d291cc8e507b12f03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 16:30:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a69512bd5504b4d84dffdb08c2f5c37 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/466a1ab9405d4cf8a63eb13ce3fc72e2 : SUCCESS in 19m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a09024296844b098eb2a8cf90e7a438 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b507c4cad187460da98e4ccbe9559856 : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d53c31b764b4a9cb84f2d271bd145ab : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/49b5f97efb5040f19948038cbf6cb59d : SUCCESS in 6m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9df7cb3050314e9194b62416bcd3a498 : SUCCESS in 4m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f672cb5efa5d4ecfbe470b69101dde51 : SUCCESS in 10m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f717f8d70b0e4edd95967f7c00c8dc4d : SUCCESS in 19m 45s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/074762c819f74c60881fc1b2b8c442d1 : SUCCESS in 11m 17s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/83fcb113740e472e8c6fa7e6436827c4 : SUCCESS in 6m 41s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c76396e912a0431fbffbbc52a0f4c196 : SUCCESS in 11m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/011014e7aa264781b6e83738a2b43ff5 : SUCCESS in 8m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0c7c1132fb274ab99fbdedce0900019d : SUCCESS in 10m 48s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/582f17f117a34d87a8069397c87fb630 : SUCCESS in 15m 14s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2fb50d9c84754640871d1a63b062c6d6 : SUCCESS in 12m 40s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f316397dea6940f985de65d540aaac55 : SUCCESS in 6m 19s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f7bc8bb0e6e84645aef144d6d0d5d18f : SUCCESS in 25m 26s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6fef7b6e742144bfa23aed0c73c3a477 : SUCCESS in 16m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/47af1b076f414c66a0704fb055c42e43 : SUCCESS in 4m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/23e83dfcb44344e8abd5ff6de756813a : SUCCESS in 23m 08s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4eb6148c252649e6b5e473a12b2453a7 : SUCCESS in 7m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1ca9d9fbb0024acbbfbe763a58877911 : SUCCESS in 8m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/44db3c59901c45888ea38a3f89b6e663 : SUCCESS in 13m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/39f75162e1df4e3f84cfddbd49b7d19d : SUCCESS in 46m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/90bc6c00945d423fb04a12ac926c5adb : SUCCESS in 1h 17m 20s","accounts_in_message":[],"_revision_number":1},{"id":"cec6512a7bde9a2f8ae6c5d16a77968fdd97ed53","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-15 16:54:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a9700f890f7614f9a7b3106b50dc89a765a1e98a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 16:57:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c7daf9214343e3bbb8ae2e1ecf4f3cd1e2656c66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 19:34:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa714dc5c12c447da1d8a3f0baf63936 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd2130d356c542e2853c07ed45370af2 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c5bd5c6bcb943e18184c7dd3037fbf4 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d73e50df37d44564be9d97a12cca7baa : SUCCESS in 5m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a977d73228814b81bb769f88f5c42ee0 : SUCCESS in 4m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2bee05858fa44a7c93f47a9f950c0b60 : SUCCESS in 4m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d6386a0aaa3a436789b3cc6203685b0f : SUCCESS in 10m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8a6530e46c4a4ff28fb43379ddb6508f : SUCCESS in 9m 35s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/eca0a6c7dfb3456b836b9235937474ff : SUCCESS in 9m 21s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0ccb8dac227e46c6b2fb995b5a5d4447 : SUCCESS in 5m 51s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0a8384321b394727a532f681179a0e1b : SUCCESS in 7m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/37138e5a61ef4f858b6ea1d52926bd4b : SUCCESS in 8m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/31c516756f7a4b14bb1a9eaa62573acb : SUCCESS in 10m 26s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b522709b66e94522b975e9d980aee1db : SUCCESS in 12m 30s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/af3f1eb6a3c34495b3b23b14ba67a23e : SUCCESS in 9m 28s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/daa6ffd896684c39aef5ed87b6a6e539 : SUCCESS in 6m 27s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/02c5900112a84eaf9511afbf919d6d08 : SUCCESS in 23m 39s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/090578cbde5d4473b92d6020b4f5a92d : SUCCESS in 13m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ed3150a598674d0c8bc4f04b5bb7c0b0 : SUCCESS in 4m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a44f14d8de73411383ee0b26f985829d : SUCCESS in 23m 43s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/599a66ded890405dba292d57ec064bae : SUCCESS in 6m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2bfd771dfb2c467d8151cd7a7d18ad97 : SUCCESS in 7m 43s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f8c2bca149a44259a96c8644d0e05e7a : SUCCESS in 11m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df0ef0c0d7724871a2d8a6b1d53c86f5 : SUCCESS in 46m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/60900c8d00ea4f12b04da3586de38a7b : FAILURE in 1h 03m 45s","accounts_in_message":[],"_revision_number":1},{"id":"2877422d90ce7d0d8df39b555d86a84a657c892f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-15 19:57:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"77ef0dd4ad5b8da73c390d14f15d8c5f4a063860","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 20:22:06.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5c3feaa0bec94c36ae3eaac63e33db96 : FAILURE in 18m 02s","accounts_in_message":[],"_revision_number":1},{"id":"4a68641437eb97751a96707a4918b0f534886cb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 22:32:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cc48cfb06b34e48b4b3b2b2b3fae656 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96f1019d22234a9eab285c90d80674e0 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0fc03c20ed7f464794925273f7f47f49 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8bdfc64d1af4717a5be556fe0adea37 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ee5a54978174c19bc7712a549da02e5 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2d20646b771f4d198a15140daa7fcd8d : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c8bf3d0f0c304f5eabe5d8ae23e0932d : SUCCESS in 4m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/84e5ad5361534364a791f3a40636c01a : SUCCESS in 11m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/943eafb796f44002b810d01843d7deca : SUCCESS in 11m 29s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3616fffd34c64aeebfaa28657af989c3 : SUCCESS in 10m 34s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5447814adb2f4b3f8a3fe8f116586a8b : SUCCESS in 6m 22s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f8127314d9634c6d8df26d45a23a0966 : SUCCESS in 12m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9860768dee524e5b9e471b9d80ec09a9 : SUCCESS in 10m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ca05d098b20c4b74baf53f5460275b9a : FAILURE in 23m 34s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b510a7169503484283b9bec17e5bb115 : SUCCESS in 1h 07m 16s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/eff98546adaa4e21ae6bb742b799a674 : SUCCESS in 10m 27s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b96000aea1c942209f4cfe730981a745 : SUCCESS in 5m 50s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9a9b6530eb7a4b08bd2c5511c7aa5a93 : SUCCESS in 39m 56s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1d649e5030244137975886bfdadc7171 : SUCCESS in 15m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f45d3623ed414a1c8cdd394db822dcff : SUCCESS in 5m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f5504328a372480ca4647610e284da22 : SUCCESS in 27m 36s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/8311f1c3c81a4dfc8972eb434df2e00a : SUCCESS in 6m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/85b291a68542436da6bd22d3cbc18671 : SUCCESS in 8m 24s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f4be0c71c26a4b0d9929ac8bf0f99fec : SUCCESS in 34m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48690d54cbf341509689d67f0788e0b7 : TIMED_OUT in 2h 34m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/09152d05782b49d7b2916483b2e28c86 : SUCCESS in 1h 33m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b5da2080c5d3c8f6520c1c60124d8c1fe0a58032","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-15 23:01:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d9a922b32413276a32183cdc3e7c97853faa9a50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 23:16:55.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/dbd83c28cb5c4eed876b42ea4c02d814 : FAILURE in 11m 27s","accounts_in_message":[],"_revision_number":1},{"id":"fccab33b1b4b1a2fc4a56269497a3547e7c9b6dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 00:38:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22a0000a69c045fe8ba50a09e7a79220 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/892af712cfa94220bf5f10e3342bbe56 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44e1a9fd8540460bba027359e95fb074 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5afb8b9b56241a8a082028125841db8 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87ee53676478437c8d5e7e85b7e876e4 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c338b1a612f649c5959bf1985d0dcd51 : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/379b769da475400aa1f875615bf1747d : SUCCESS in 4m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b57eb18466b149e8aed8b97a11f8a18e : SUCCESS in 11m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/57f04406aab442eda3bddd24074f6a57 : SUCCESS in 10m 22s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0a2a52d71a1c4171aa708c05cf401756 : SUCCESS in 9m 24s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c4d220fc8c174a0d923120f2aedce367 : SUCCESS in 5m 32s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b4c6f31ed4e4485282bae9e553df91d2 : SUCCESS in 9m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/34dcab07c1b841d5bddb97761732f8e9 : SUCCESS in 8m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d460e9a5d4d540cfa093c873edfa7e6e : SUCCESS in 10m 31s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/73891f98ca6b4bea941c3a2ff797d4c7 : SUCCESS in 14m 43s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/80f8d8877ad34b2e95e993bb06551529 : SUCCESS in 8m 59s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e96eeede4446400aa1f44b7cca5c52fb : SUCCESS in 5m 02s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c6a09e1751e8447a8573ff05df5a7dca : SUCCESS in 29m 16s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/61991c95f05946649ef6b4cfddc2208c : SUCCESS in 13m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d62243995fd94158b99da7d787174ab8 : SUCCESS in 3m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b28939a3612b42e984074a877cdd013f : SUCCESS in 16m 55s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5739d3bd20414c649245b52e7e3b21ca : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b898ef4310c34237bfed4d6c80deb01a : SUCCESS in 6m 29s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9e6130742e274fe98afd60300b423d47 : SUCCESS in 13m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9f08ea8ef11464890226ee7e5d45590 : SUCCESS in 48m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7fb9aeeb2ae5482d855e61e570e6cf7f : SUCCESS in 1h 30m 19s","accounts_in_message":[],"_revision_number":1},{"id":"7a5575f561d69af78bdf90890bd27b2543987d39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 00:38:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c205cd2dd90189e4110683ad23d629ed57106422","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 03:29:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18c64072c6414975b2239c4d2e5e5978 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a10ce9307234941b43e23580a73e994 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e10358b2088b40969680f2df78bbd573 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e676f81f4cc74e7fbe6cc2275a3d34a5 : SUCCESS in 5m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9ae9d082ee784304955a3a553b180ac6 : SUCCESS in 4m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/62c2e51cb278411c8e7cd22bee161f77 : SUCCESS in 4m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a6212fb713a545da8127cc4a4f41ad5c : SUCCESS in 10m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e545273e9b994eae9ba27de23a476e0d : SUCCESS in 10m 56s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4e08d97ff64c4a7a8ad48e521d2d9f40 : SUCCESS in 10m 00s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d629af7ff4f64a0e9511f74a761a3a58 : SUCCESS in 6m 14s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/16e9c0fce8834550b4fe8e8997935198 : SUCCESS in 9m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f7ba829fe41849bab594c2c661c6cdef : SUCCESS in 9m 23s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/af9cbc75ea52423fad7c9de04ed584fb : SUCCESS in 11m 00s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7b30fd3b7104429bb1c404fef5cf512e : SUCCESS in 14m 18s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/877a3195702940678bbd8fb18a7914fb : SUCCESS in 9m 46s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5448b69c6f80458f88299d97fbbc5ac0 : SUCCESS in 5m 41s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c58eb1b7348645cb984fb7897369177e : SUCCESS in 20m 23s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8ff97167858e4a378abe89622606ea23 : SUCCESS in 12m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/379a8e6507984db8865e0f0fe2d3cacd : SUCCESS in 4m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9079a15579ac4fc885ec7a219b1f2897 : SUCCESS in 17m 47s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/dc1e34f8704e428fa32a5f985c1e80ea : SUCCESS in 5m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/95c6bdddf83a4a8b8a040c4aef9aefeb : SUCCESS in 7m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/90292ba7b0ff4f238cb2893bc55f0563 : SUCCESS in 14m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a68f4592cbdc48918a5791aa85b21def : SUCCESS in 44m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6229c61a3d6345f78ad28857acb2f020 : SUCCESS in 1h 15m 55s","accounts_in_message":[],"_revision_number":1},{"id":"040246f806a49f6c6a6df0889c6aaebc4d9fe96f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 03:29:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b702b88b947572e514e4e59709663063b5a98ab9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 03:30:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1507a7e25e7c4526951c02645111b0c2 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"31aef2de5e48804dbfe79e018189092b914e519e","revisions":{"31aef2de5e48804dbfe79e018189092b914e519e":{"kind":"REWORK","_number":1,"created":"2021-06-15 10:06:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/16/796416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/796416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/796416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/796416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/796416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/796416/1"}}},"commit":{"parents":[{"commit":"e15a2a9638a806899b5a507e16aac57380562ba5","subject":"Merge \"update constraint for tripleo-common to new release 16.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e15a2a9638a806899b5a507e16aac57380562ba5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-15 10:06:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-15 10:06:06.000000000","tz":0},"subject":"update constraint for oslo.db to new release 9.1.0","message":"update constraint for oslo.db to new release 9.1.0\n\nmeta: version: 9.1.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I5ba11909a1bc08c7262042df172a13ff5b215199\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Iba80669226464500778b9d2048252bde0a55145a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/31aef2de5e48804dbfe79e018189092b914e519e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/31aef2de5e48804dbfe79e018189092b914e519e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
