)]}'
{"id":"openstack%2Frequirements~796647","triplet_id":"openstack%2Frequirements~master~I2d5a7dfff31fb0638a9e82948052a73a4db6fa85","project":"openstack/requirements","branch":"master","topic":"barbican-crosscheck","hashtags":[],"change_id":"I2d5a7dfff31fb0638a9e82948052a73a4db6fa85","subject":"Add barbican crosscheck job","status":"MERGED","created":"2021-06-16 12:44:08.000000000","updated":"2021-06-17 10:48:06.000000000","submitted":"2021-06-17 10:46:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796647-barbican-crosscheck","meta_rev_id":"20fb57d036c7e71d2111c907b2ba4fab2b28a60f","_number":796647,"virtual_id_number":796647,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-17 10:46:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-16 18:17:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2021-06-16 13:28:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-06-17 00:46:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-16 13:28:57.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2021-06-16 14:37:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-16 16:32:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-06-16 18:17:29.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"4fd1a7ae37401ca2dee67f9defc6381448e0c32c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-16 12:44:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e03cc7c95353c13c93426ba7ca7cca8b479745c0","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-06-16 13:28:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"54985d971bf16ebd663d8083e5c88a8c1760348f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 14:37:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/469df9382d94463f8e1c77e40596514e : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/401d3c5b28fb420b83996ebc9640b6e2 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a4fb67e331847a7a8d247cb26a73f65 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/505d5b7ee89f4b51aa05e219e950dac8 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8415e22e0b494d23b41a35747a3a9290 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ca8f226c38994bb7ae2175b914142c08 : SUCCESS in 4m 06s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/fb59fef4d17248c2b4905e933722605a : SUCCESS in 6m 20s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/fd490fca897f406f824eed8aba204876 : SUCCESS in 9m 54s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a278ca6b56764ee7bcc4800fe59dc434 : SUCCESS in 6m 38s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/48bb353f95214c72a53d08d3de06ac6d : SUCCESS in 9m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bd635acd0bbe4f31adb9ea83507f7d0b : SUCCESS in 10m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f0a541aaa6c24fb892225505030a646f : SUCCESS in 12m 04s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/db341aa41ace457682b596beafca3eac : SUCCESS in 17m 10s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7ae67813823847629de71ddacc07afc9 : SUCCESS in 10m 57s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/23e890c10d214433a09b1910c2133e1b : SUCCESS in 5m 31s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7c88abc76f4a453d9a2f584ed17b95a9 : SUCCESS in 24m 57s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/96adc8bda4b84e5bbb3ed7e6bf82632f : SUCCESS in 14m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8881f8614236428fb6b23f50bc92735b : SUCCESS in 4m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2a756cd9d5ea4c00865edccd36795957 : SUCCESS in 21m 57s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e2dcd8d64d974fdb98e22d1970c62dcc : SUCCESS in 6m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/88d49ccfc7584480af7782b48e1eb6ee : SUCCESS in 7m 27s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6dd493f5209645309a672e33a7e8a284 : SUCCESS in 11m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f94e58a08c84b5c998fc78b549859a8 : SUCCESS in 51m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/def804e8164845b399334bb8bc6be5b3 : FAILURE in 1h 09m 30s","accounts_in_message":[],"_revision_number":1},{"id":"19aa234ca12f5633f235e661d67df955ba4dfcc9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-16 16:32:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"89d81b85c7a0d74f7c42a1b43b70d45ec4e10b38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 18:06:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df06808f980b41b0ae5f35a7df1d843c : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d99b74254134a6a916517f203a54496 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8218c2ebd264431ae3b20af0ddc710a : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77f1569f7449402aa2b3e50dee298d8e : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d1cc4ed384d4b1eb7a9ffcdd548ce34 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/be444792d59d470fb1bc2033810e73b9 : SUCCESS in 5m 30s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/8e88f624308047149494f1d8e293cffb : SUCCESS in 6m 13s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/36c61e5c15344b458c39a2be4350ce15 : SUCCESS in 10m 30s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5d34282298c84b9198937d2d745e1f7c : SUCCESS in 6m 34s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/01ac512dd84e4f3b81538a30409aee39 : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c01cf2e84ef940a89f39d62460938641 : SUCCESS in 9m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ae0ae7df341640a3b5ef55bdc2d9cbf7 : SUCCESS in 11m 57s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/cbeb69f95e914520bbde72c2e15b3368 : SUCCESS in 19m 19s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a066f11e54f940fb80c3a9b6aef42b2e : SUCCESS in 10m 41s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c33cc5c22b5c434e9077082a5562606d : SUCCESS in 6m 22s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1a36e3d2af524aa690afb4e00e75a59e : SUCCESS in 26m 41s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/57dbaec87e234055b6c28662ee9374b6 : SUCCESS in 13m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/21a955b8dbee44f7bd74a9883985eac5 : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/659b6880d94b4356a00ad6e6611739df : SUCCESS in 21m 33s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/08dc15e318924a8d9984f48102a27495 : SUCCESS in 6m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3bbee76dfe1a400880520ee7b022c28c : SUCCESS in 6m 36s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6700a5cbd2b6490ab5a7dcf9e475a399 : SUCCESS in 14m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/29b9743b02fa49bb869157f3259f89f0 : SUCCESS in 58m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/49c79b5495e240d9a642eb91738e1c8f : SUCCESS in 1h 28m 07s","accounts_in_message":[],"_revision_number":1},{"id":"0dc68261d8e4ad72c6e375fdbf8f8226a209e4b9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-16 18:17:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"eff1d39d7c0c5cc4dcd298c5edb00e52a1045609","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-17 00:46:38.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0b566063671c8a5426736f7570b44e87b7e6c5be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 00:47:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"485cf8d40197ff70b52803c9fe1c130099c21395","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 02:23:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb54fc2fdd984b7b9789dfc5e4c1be70 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/830c10d63aaa4613b9bb35d779551d0e : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f69c761f07d14823a6975a83b4e06077 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5917dbe4e3d34a7287b9bf27c86b95c2 : SUCCESS in 4m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4b9ecec267bc4dfebe8397c28e9cd72e : SUCCESS in 4m 13s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2e40455cb8244c72931e20f1250d788d : SUCCESS in 10m 46s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/955fedeaf8184284a8ad9a39e23959ec : SUCCESS in 5m 50s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f018c70026f0409bb341b1d0c3aa8f07 : SUCCESS in 11m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/28dcacd78a64454084ac1972bbab979c : SUCCESS in 8m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c97030e671c2456ba1d211bbcb2b9dcc : SUCCESS in 10m 48s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f66ace7fb6554297a1d1f9024602af8f : SUCCESS in 14m 42s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5f731162cc594a7dad454fecf7167098 : SUCCESS in 14m 34s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/4a77098605164804bfe0c3b43c7a803c : SUCCESS in 9m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/fcca265b60ef4041a287a9a095538626 : SUCCESS in 24m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1cdb2aded4ff4eb98962d3859ab610f3 : SUCCESS in 4m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/778f3da62ab746c39242e42c8eb20a94 : SUCCESS in 20m 43s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/8f2708b2edcf474b9080934dc765cf92 : SUCCESS in 9m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bbc3e8aef6e7473a80b74d263bcb67d2 : SUCCESS in 11m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1c869bc78eef47c2aacfccf1641b1e42 : SUCCESS in 12m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f8c5ff29bc5427d84120bc7ce1cbc34 : SUCCESS in 51m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7c1bd922783d4b56a871af2ac0c6ee41 : FAILURE in 1h 34m 28s","accounts_in_message":[],"_revision_number":1},{"id":"014ff9aae1ba9c6ac66705100a2900f5568c1276","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-17 04:33:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"22563e6ea50a475707cc87c01e2e53e091b02dad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 06:03:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dafa0eefc12a494998215e84bd788d4b : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/feeba1207cfd4129b8800fd43f6e90b8 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db486fafec0d4ac0ba2a0a72f414dd59 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64f02bf9c0a94b0997243c7837fa7f35 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8280ad7f41354ba788d1e6f0924898e4 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9f086713bbe645cb9817c15ad8a975c5 : SUCCESS in 4m 31s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/024c9a776cc048b5895a28a64d315770 : SUCCESS in 6m 18s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c787322114014266a8967f165bd45e25 : SUCCESS in 9m 49s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/003ae4d90cab45408384db1f682a439e : SUCCESS in 6m 53s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0b4096fe99c746cfabd56a305a55eedf : SUCCESS in 9m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f5ec56e0d56349c4b607ab0c95a50159 : SUCCESS in 11m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/94662fd633a44a79afa6dedf4021cb98 : SUCCESS in 9m 44s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/fc6ea5d04e384b759b9f6bc3155bcef1 : SUCCESS in 16m 37s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b00c21325b4249a79fd78e628c49d8e7 : SUCCESS in 12m 40s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/80cce71fe16241ce92547ed23a0a2f44 : SUCCESS in 6m 43s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/cd984687dd5f486e8ec5ff081230a85f : SUCCESS in 20m 21s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1f3478c1fd8c46f7b973cd18d8415f7a : SUCCESS in 12m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a4374c8f35464141be487ab23e8dcf07 : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7e154f249a54413c8554854d0f8eb2aa : SUCCESS in 22m 23s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f9155c01dcf14e9c98270bac3a75aab6 : SUCCESS in 5m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0b1e8640a24345968307a41ce47c0859 : SUCCESS in 6m 47s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0ebfcd6c65cd4a05963c7c054036d17f : SUCCESS in 12m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87f8892d2808468eb38630eacaf6f101 : SUCCESS in 56m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bdc2b13d48524b1d8000b509f20a032b : SUCCESS in 1h 24m 11s","accounts_in_message":[],"_revision_number":1},{"id":"3bf8a876adfa457130452902952e1d525a295ed2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 06:04:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"620938ed563b33616eb7a84b8a6f0080deb2afd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 10:46:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adae8204599545e0a083ac7c16338dec : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0e8f2494db24882b9065fdfc0cc429d : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ebcc5cdecd54794a5b394f1ef40b3f5 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78f69c7af18a4894b23f97735a64cda9 : SUCCESS in 4m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/33f78fc10f6d48d7ab6bcc63889585c1 : SUCCESS in 4m 35s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/922780ee9d3c4c8cb69be7731293dd28 : SUCCESS in 10m 42s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/069f0523aeea463481e4b9e552d9b41b : SUCCESS in 7m 37s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/72139bca7e2e4dd6931962dc50c7ea66 : SUCCESS in 12m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d8de1c12c78541db88702193421dd389 : SUCCESS in 8m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a3f1c0ba79b1416fb7ca2c3ba390795a : SUCCESS in 13m 09s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e5ba80192fce49b3af49266c884d145b : SUCCESS in 14m 18s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f0f82f013d7d4f49b6468b12dee943ac : SUCCESS in 11m 25s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c36f5a330a654f6684d2ece2d25d4a70 : SUCCESS in 5m 58s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8f8e1f01a3434c4cb6ec3d594de7411f : SUCCESS in 25m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/18d8a25131604247bfd55d3f15cc9dbe : SUCCESS in 5m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a9dca07f2f8e4ac5b893e792e44c1a58 : SUCCESS in 19m 37s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/627dd528fe4f4770a6ec9a184c1c3cbc : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f878611b28b04a86b97ddbf3d432bfcf : SUCCESS in 7m 54s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6e420c6df5cb454182fc461418c15280 : SUCCESS in 12m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5d1cd738e48a4330aed8822b6ac6a7dd : SUCCESS in 48m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6e16cf350de9406bb8a78a7b020a0408 : SUCCESS in 1h 23m 54s","accounts_in_message":[],"_revision_number":1},{"id":"61d13e4a5b9712cf5995a554a0bd3dcf3ef03859","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 10:46:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"20fb57d036c7e71d2111c907b2ba4fab2b28a60f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 10:48:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a783c56c4d2a4f0eaae993c2b8312551 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6ddfe19bc5b0745817e42a38f7b25adad00ed903","revisions":{"6ddfe19bc5b0745817e42a38f7b25adad00ed903":{"kind":"REWORK","_number":1,"created":"2021-06-16 12:44:08.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/47/796647/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/796647/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/796647/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/796647/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/796647/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/796647/1"}}},"commit":{"parents":[{"commit":"1cba0038a3fa21082492a9c9e1ce7399eefb5986","subject":"Merge \"update constraint for oslo.cache to new release 2.8.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1cba0038a3fa21082492a9c9e1ce7399eefb5986"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-06-16 12:42:20.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-06-16 12:42:20.000000000","tz":-240},"subject":"Add barbican crosscheck job","message":"Add barbican crosscheck job\n\nThis should help avert some unpleasantness caused by, e.g.,\nchange I1b8df63bfbae.\n\nChange-Id: I2d5a7dfff31fb0638a9e82948052a73a4db6fa85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6ddfe19bc5b0745817e42a38f7b25adad00ed903"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6ddfe19bc5b0745817e42a38f7b25adad00ed903"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
