)]}'
{"id":"openstack%2Frequirements~796827","triplet_id":"openstack%2Frequirements~master~Iced7502896a2a30942cee3927a83235893cdf9fe","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iced7502896a2a30942cee3927a83235893cdf9fe","subject":"update constraint for ovsdbapp to new release 1.11.0","status":"MERGED","created":"2021-06-17 12:08:12.000000000","updated":"2021-06-17 22:37:44.000000000","submitted":"2021-06-17 22:36:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796827-new-release","meta_rev_id":"3702ee50380af3181e0b3a57d539d66ca2969a7a","_number":796827,"virtual_id_number":796827,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-17 22:36:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-17 14:46:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-06-17 12:20:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-06-17 14:46:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-17 12:20:00.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-06-17 12:27:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-17 13:24:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-17 14:46:23.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"0ac8e4c70630b702d8a98dbb337eb568ad0776cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-06-17 12:08:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f0315ab0e94be1835be0ceb380efd9e5ca0d91f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-17 12:20:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"94281bebfdc762d55a85a6538367c6cddc296c30","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 12:27:49.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/90108036469042cda927628e2b5ff2a8 : FAILURE in 17m 52s","accounts_in_message":[],"_revision_number":1},{"id":"e1825af8a94a9fea3e8f5cde61f5c80eb6398ecf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 13:24:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3ed335c1a954c9c8baa06a1a2cb916e : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e47dc06a92774f71a6fa06def3464b02 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb7b2b4a0ba548cfbc9805df3b24ef80 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b2fed0bfc924272a3344074498f5000 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e71caf5a0bd40f1888c30975f2d8134 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fefff13af5694bb7af346fd10e447263 : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c7efabc9464a4587a9fc8de2125ce46f : SUCCESS in 5m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8cdab31ebeaa4e3494b63c2b368ea069 : SUCCESS in 9m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8d46905315ce45479cc1a7738e0a4b70 : SUCCESS in 10m 47s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/4de309041d9a4efa96b4619bb875ba47 : SUCCESS in 7m 16s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b3d770bb86874f1b872bbbbd91e791aa : SUCCESS in 10m 53s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a6dfea0e80e64fec9a68153c0a2c49f0 : SUCCESS in 5m 53s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b445726d726d45d7b24293941466a090 : SUCCESS in 9m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e5493c1860124de6809e594fff9861f7 : SUCCESS in 8m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a7af0ed85271458c9f04cb680e3986db : SUCCESS in 13m 49s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e4b94176d0f742faa11b13f8e0c62770 : SUCCESS in 12m 39s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/92faa829eea341a08099acb322c365dc : SUCCESS in 9m 05s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8a6878d86f324304995a0adab12945b5 : SUCCESS in 6m 04s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/aae02064748b45a5a5c771516709b533 : SUCCESS in 21m 42s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ef083cd0dc014710bfc110e45d1b7b1d : SUCCESS in 12m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ad8b51f370024add8f40e37e5d1352f4 : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2304f13b97a7421c98ee447252768e7f : SUCCESS in 18m 22s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/591ab137efd041e6a1e2afdf1350616d : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2f02257fb32b4369a1b7e78ddb15dc35 : SUCCESS in 7m 17s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/54cd194e5d8c4527a6eeae93e78bffb6 : SUCCESS in 10m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b6e766f7f3e0449485020f72481eb18c : SUCCESS in 42m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e2eef6988eb84567a2c456fc1d98120f : SUCCESS in 1h 14m 23s","accounts_in_message":[],"_revision_number":1},{"id":"ad874b07b7e54d80269b466896ba84f0d3112c7c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-17 14:46:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"33c52d574e56583122e849ec2c3ebf5845149e93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 14:47:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2aa996e928f0c74db0aabdaf23a1a19b039055b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 17:30:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f48af441ff0e4481ad0a716c577fd2ce : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/944c8d09552041e3b0c347477b1c4826 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92ceabdb0b9c443fb797e4981510a468 : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10cedc53ca94441cbe28a0d49cca7faa : SUCCESS in 6m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/424a3c3fc8fe48dd81bbc6163d072650 : SUCCESS in 6m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/59630479f61e4baf91c96ce958a73ade : SUCCESS in 6m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cce171c4cf1f40ee828e5240d589f2c4 : SUCCESS in 13m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/24121880a5d947ff90ed683180c3d870 : SUCCESS in 11m 40s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4d5254408f1743eda7c76f6e3f891286 : SUCCESS in 11m 21s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3bccae3138114e9cb152473d5062b841 : SUCCESS in 9m 07s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d664a579db48462c95d8ba202dc3b73c : SUCCESS in 11m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/285ec8e4450d43d6818146bea82c68b1 : SUCCESS in 21m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4cd374982d2543f883c0e6ebc39a5a7c : SUCCESS in 11m 00s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f65e43e212d74ffd8f4c18d67d364746 : SUCCESS in 14m 39s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5bf903f616f449308447928857cef8a3 : SUCCESS in 11m 32s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/73173321872c44c99d28b1a87d7c8d8b : SUCCESS in 7m 24s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b6b2265a19da44218c571df65552d9a0 : TIMED_OUT in 41m 34s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3fd2d6032de6408d9a7e1584766b4f4d : SUCCESS in 13m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5cdfb4ca47944a1db834644e943f1052 : SUCCESS in 6m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b3c8190b4f34449291b651e663b02da0 : SUCCESS in 21m 06s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/582ca5f48f7b454b9857c63bd48006b9 : SUCCESS in 7m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1e97f60788b24cce80dafb2c41313706 : SUCCESS in 8m 10s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0045186fa3064d4c8d65db89a38dd97e : SUCCESS in 13m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6eb703f3c3614e4f810d1d9f20a945ce : SUCCESS in 50m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bcaa4d03ec644c998e22e9494333d426 : SUCCESS in 1h 16m 44s","accounts_in_message":[],"_revision_number":1},{"id":"e7346c3220c8fd833103c221ab72d47230090d3d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-17 20:03:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6d8b3b129998a6339eedf304b682410b7e9082b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 20:23:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/220342ad55c4430d9922a0bc9aa91963 : FAILURE in 15m 10s","accounts_in_message":[],"_revision_number":1},{"id":"3f596c82a7127cc5a55cdc5847a87ac2bbb415c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 21:20:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f359c631ec249fc8823d8fbe74e49b4 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77dc230b4ada454eb9fa9b980daa71b0 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/877c341edbc34cf899ec5e1f9b73109a : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f5bdb5018104f7a84cb0a0bafa3022e : SUCCESS in 9m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cc9deefddd94fa1b80909e8c76ec16c : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7776ec4082d441f38524143db4fea2bb : SUCCESS in 4m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2c923f5f54f54145b234cc7c89eba2b0 : SUCCESS in 3m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/75e9f40da85b4cbb9a7f5cb97b50e4bb : SUCCESS in 11m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/eaa1e87f59184f42b1a96048adacc2d1 : SUCCESS in 11m 04s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/c3bc88e9b4624ad28b7d0f4b800bcc07 : SUCCESS in 6m 38s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/cf2b413720e94cc49082fe07ebd31f38 : SUCCESS in 11m 02s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2b96f96e23c34cf89036556b8e93d1c6 : SUCCESS in 6m 16s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/970b338b631c43f581881c9844f9204a : SUCCESS in 9m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4cfe441090304fb0b1f5265a5abb00f4 : SUCCESS in 8m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3dd7f3fc5d4845e79582b3d2c014b2d5 : SUCCESS in 10m 03s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4795e23804ca4863a0346283dbda162e : SUCCESS in 14m 50s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d0cd2793853144f4b4f1f74ee3f9a4f2 : SUCCESS in 9m 15s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2abdd2624cb248f0a8cd313ee4f41a12 : SUCCESS in 5m 56s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f79bcc3435844e93b8c67ae12549c1e6 : SUCCESS in 20m 07s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a96910be233544199a44b0439a7d093f : SUCCESS in 12m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9b63b07b1dbd43ac847611d468ca3ba0 : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/307321a3e67044379ad6266d13d54b8c : SUCCESS in 19m 03s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c1803a10a87546d0a50acd265cdd4963 : SUCCESS in 6m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3e2d51e054a549e19c9ea3e4f67ef733 : SUCCESS in 7m 13s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1200c327014e4736ad4c4a09641ee0aa : SUCCESS in 11m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/99bd650bd2f244e4bb8f74756de3820c : SUCCESS in 43m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c6c0a7015b1a49b6885e54d5c92b0f6d : SUCCESS in 1h 13m 22s","accounts_in_message":[],"_revision_number":1},{"id":"e295825a596eb640db287fcbd8d89e8875b5ff2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 21:21:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5de26df9b4b6d8579dcd9491c189d92c3dc01209","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 22:36:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98dae789440b4823b8a5cc807ea1b4a5 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62bf31a5e9b84877a5e8435a21cd7144 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e9c8dd64431433e9c59881e101e58c1 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76fc844310b64e1d9608ed0839912ad4 : SUCCESS in 8m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f9de1599c83a4d32ad5d69f0b13f7738 : SUCCESS in 5m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/75a2ba03746846f7bf638eb769544454 : SUCCESS in 5m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cdc6253448a24435affdfa8c19f8913e : SUCCESS in 10m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a14b31152475444d99a15a1b96115832 : SUCCESS in 10m 24s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/126f017affa048e6a03a9427fc595bd9 : SUCCESS in 10m 35s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c07a8bf16214411b83aaf531109678a8 : SUCCESS in 6m 54s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6e637c23a869414998fc81e762714c45 : SUCCESS in 9m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/be08df56ef35420e81a02200e83508cd : SUCCESS in 10m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/535349c63ac9485ebc60816849c18c57 : SUCCESS in 11m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8c63826c1b264fd5ad76d357843511e4 : SUCCESS in 14m 56s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b8ad8607234e4aacbec61bc79b7dbbbe : SUCCESS in 10m 19s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/3ad241d637bd43ecbf5b3769c2b7f842 : SUCCESS in 6m 17s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/50f405ca4e8449698f90c057bcf367fb : SUCCESS in 24m 58s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/73b5a1fa68174347830454cde89bfdd4 : SUCCESS in 12m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/897845905f3f4e9a978db215a475527d : SUCCESS in 5m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/59dc432e794a4487a132487e08d7f9d5 : SUCCESS in 22m 33s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4c86652f98f344cf9309c2915b7cf4e3 : SUCCESS in 6m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/07b7485f29a144ff9773590863b43dfb : SUCCESS in 7m 58s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e6e51a65139a482ab2aaadb7d678534c : SUCCESS in 20m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a30449a3ca1b43d295dfcc693b6ca8a9 : SUCCESS in 45m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b1e6823f87dd4bf0988dcb5deb225af6 : SUCCESS in 1h 14m 09s","accounts_in_message":[],"_revision_number":1},{"id":"d2a7762cce131de9c7e6ea85b0d71c982cb20969","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 22:36:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3702ee50380af3181e0b3a57d539d66ca2969a7a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 22:37:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37a1d3f86da84f73950e54bd0277623d : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3a317a797d346bcc21340a4ca44d52716099f5e1","revisions":{"3a317a797d346bcc21340a4ca44d52716099f5e1":{"kind":"REWORK","_number":1,"created":"2021-06-17 12:08:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/796827/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/796827/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/796827/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/796827/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/796827/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/796827/1"}}},"commit":{"parents":[{"commit":"6ddfe19bc5b0745817e42a38f7b25adad00ed903","subject":"Add barbican crosscheck job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6ddfe19bc5b0745817e42a38f7b25adad00ed903"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-17 12:08:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-17 12:08:10.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 1.11.0","message":"update constraint for ovsdbapp to new release 1.11.0\n\nmeta: version: 1.11.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Ihar Hrachyshka \u003cihrachys@redhat.com\u003e\nmeta: release:Commit: Ihar Hrachyshka \u003cihrachys@redhat.com\u003e\nmeta: release:Change-Id: I60a41cf6097e5f8bbc008bc61f653b6a3abdf361\nmeta: release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Iced7502896a2a30942cee3927a83235893cdf9fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3a317a797d346bcc21340a4ca44d52716099f5e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3a317a797d346bcc21340a4ca44d52716099f5e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
