)]}'
{"id":"openstack%2Frequirements~796828","triplet_id":"openstack%2Frequirements~master~I3d9ccaaef54759cf820b4dbef9c64d2317b564cb","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I3d9ccaaef54759cf820b4dbef9c64d2317b564cb","subject":"update constraint for validations-libs to new release 1.2.0","status":"MERGED","created":"2021-06-17 12:12:43.000000000","updated":"2021-06-17 18:14:36.000000000","submitted":"2021-06-17 18:13:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796828-new-release","meta_rev_id":"a495274105a91483a62cb6acf9d2b14554845344","_number":796828,"virtual_id_number":796828,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-17 18:13:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-17 14:46:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-06-17 12:20:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-06-17 14:46:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-17 12:20:28.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-06-17 12:25:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-17 13:39:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-17 14:46:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"7bba7919b1b78d78592d79786dc02b57123a5ee6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-06-17 12:12:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38f519394af7cc0d66d9b847591ef61ca9974bbe","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-17 12:20:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"df7f78099ba48f640e6c30dd5a43292f04e60505","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 12:25:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/59aff798193f4224bbd3b48bd0663cb9 : FAILURE in 12m 15s","accounts_in_message":[],"_revision_number":1},{"id":"ab5e5aaf0c2287eb5339d8b4b8b6b11b01c0f2d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 13:39:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c547336dda854508b9d2ce0c1f18e2bd : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77d174901b824d08b9a2bb59bd8e4ee6 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0225f6e6e04b49d7ab0506c7e402b203 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3c4a255537d48ae81980f2987891b86 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e526f632daa4f21a286b526a5f0e8a1 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2cd59ffc672b4dd8993ac24814335637 : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c6d66dce63f14aa989c5ca76389f3222 : SUCCESS in 5m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/77bccd4afef544eab1e05e4077653fcd : SUCCESS in 16m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/67a9b090ecad4d5d935bea737545d92a : SUCCESS in 10m 56s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/1a3a9c65139148bc9e870290a3fbc893 : SUCCESS in 6m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/11a709d2240444e5aa4d429010533f29 : SUCCESS in 9m 59s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/fb78fdcd5da44e56b9f7449aa7af7318 : SUCCESS in 6m 00s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a4661177c11943a5a04a5fe27c50ad9f : SUCCESS in 9m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/291b9c631d4c41279ca9e1d98e633f46 : SUCCESS in 8m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/be61f31f19624100b6c9b76fb1b33aca : SUCCESS in 10m 21s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8af7e88edee744bcaa4057c1542a98bb : SUCCESS in 14m 03s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d28e5412ed614294acb7a4c4cd0069fa : SUCCESS in 9m 02s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8cedc6f2af58481e9e8b8b3b533aed60 : SUCCESS in 5m 19s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/55b865784ec94f96ba68cb31d447b991 : SUCCESS in 23m 27s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/44f67071dbaa404dbdcb1f1f5b98b82d : SUCCESS in 12m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/88d27d5afef44cafb2f18e483d2d4d63 : SUCCESS in 3m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bc7ad5c7cf4e457e80821b36a55bd58d : SUCCESS in 20m 20s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9d118626a4404219ba16f36dc01840e8 : SUCCESS in 6m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/77e7436b688b4aa2bb18ccbc786837b7 : SUCCESS in 7m 00s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1a4d5118acca48e7a673d9eed9d5e8e8 : SUCCESS in 14m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/07e88c436b7746708b437c80c6f90e51 : SUCCESS in 45m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e3781fa16a3240d28963d75822bdc0fa : SUCCESS in 1h 25m 00s","accounts_in_message":[],"_revision_number":1},{"id":"fd2ce201853150ce0a3f888f4d4d4860ab6ad508","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-17 14:46:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2d633382494f78a7c167a0860f9e0a415e9bf294","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 14:47:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"db07de26329a4cb14cd156ba70cba214c66d6426","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 18:13:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ee8588dbc3f4d7e8f88cdaad15b7747 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07867d37d0f84c05afd89cfa5063ac53 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bba2ca2713af42d28925b18c3f0c38cb : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/faac560826224e5393f7e44882a34cc0 : SUCCESS in 5m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b4e7fa1b3131485ab615fbaaa6a681cb : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1a124bf276c841b5b8a61d87026ae705 : SUCCESS in 5m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9b575c44bafc43b18be894d58bc42a63 : SUCCESS in 10m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fe3b52d8055e462c9b8f0609fb1d8636 : SUCCESS in 9m 41s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e773f59ece00449ebc02719549c1c19e : SUCCESS in 11m 45s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ad4b4600ed2e4693b32e012eeced83cb : SUCCESS in 6m 40s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f7820c16dbb3478eabe8679aa598b965 : SUCCESS in 9m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4f59e97921dd46ec88a70863f9a612a4 : SUCCESS in 8m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/51f58a4a3e794766a1bb90c0a3ab0650 : SUCCESS in 11m 09s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f02e133639b146f2b0753f9133d216e1 : SUCCESS in 19m 25s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a5537467f9a643c68954b86ee6ab0c0a : SUCCESS in 9m 17s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e2b335d026ab4c4ba7e85f9ed63e1689 : SUCCESS in 5m 42s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/030c673d748844ba84911797d5d2ea06 : SUCCESS in 24m 12s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/fd52de2dde0549acbb8f5bba6691068f : SUCCESS in 14m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8186d27a311e4a918866dc10eef3c429 : SUCCESS in 4m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3a502aa53977498189de900909621512 : SUCCESS in 23m 08s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2777765b291241778dc266793d0646cf : SUCCESS in 6m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/375becd06edb4d5287dbbd0146793d28 : SUCCESS in 7m 33s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e41dd1c03fca4623a166c06ffe66d239 : SUCCESS in 10m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4865081ce86049c998ebdf8357435986 : SUCCESS in 47m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ad5e6e2f00c24dd6b921f5c924fd790d : SUCCESS in 1h 14m 16s","accounts_in_message":[],"_revision_number":1},{"id":"638811f166e39f78007e67013e7441b363406c76","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 18:13:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a495274105a91483a62cb6acf9d2b14554845344","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 18:14:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dfbdaee6d9f49b29658464a7a789073 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"efb624d26d90ee68abf087163f55061f3c61762c","revisions":{"efb624d26d90ee68abf087163f55061f3c61762c":{"kind":"REWORK","_number":1,"created":"2021-06-17 12:12:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/796828/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/28/796828/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/796828/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/796828/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/796828/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/28/796828/1"}}},"commit":{"parents":[{"commit":"6ddfe19bc5b0745817e42a38f7b25adad00ed903","subject":"Add barbican crosscheck job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6ddfe19bc5b0745817e42a38f7b25adad00ed903"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-17 12:12:42.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-17 12:12:42.000000000","tz":0},"subject":"update constraint for validations-libs to new release 1.2.0","message":"update constraint for validations-libs to new release 1.2.0\n\nmeta: version: 1.2.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: matbu \u003cmat.bultel@gmail.com\u003e\nmeta: release:Commit: mbu \u003cmat.bultel@gmail.com\u003e\nmeta: release:Change-Id: I5706179cc77d3c8dce98d4ccb43d45df81de6687\nmeta: release:Code-Review+1: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\nmeta: release:Code-Review+1: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I3d9ccaaef54759cf820b4dbef9c64d2317b564cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/efb624d26d90ee68abf087163f55061f3c61762c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/efb624d26d90ee68abf087163f55061f3c61762c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
