)]}'
{"id":"openstack%2Frequirements~796993","triplet_id":"openstack%2Frequirements~master~Iecc516fc4be310f5bd85d46a00b596801c96562a","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Iecc516fc4be310f5bd85d46a00b596801c96562a","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-06-18 06:34:39.000000000","updated":"2021-06-20 23:30:36.000000000","submitted":"2021-06-20 23:29:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796993-openstack/requirements/constraints/noclob","meta_rev_id":"6e4f3bc0a1ec2f3d52c1c0a0d0637f5599371944","_number":796993,"virtual_id_number":796993,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-20 23:29:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-20 22:21:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-06-20 22:21:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-18 06:49:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-18 08:27:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-20 20:40:38.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"2da2e1e48b7f01cb0a0836e09520ee5c85a7365c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-06-18 06:34:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52ac1fc1900644cb69745ad9c34774d7ea734469","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 06:49:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d5aed5f720044dcab17aacfa44252059 : FAILURE in 12m 43s","accounts_in_message":[],"_revision_number":1},{"id":"19290298e72b6f22f47f0bf1973f5e116e433f59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 08:27:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44fee8f3887f4511af8fade0e49f5278 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/481637fbbd274be0a65e3b272c7cd315 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbdba3f5ce8e4d78a8122f9e620ee13c : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25869ba295a447afa940830168347564 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5c6c69e94d6446390c6d2341c98dc9d : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d026c2375e50497cb92feace58d4cea2 : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/09c5ba2d21b240848ed36e62534e5f40 : SUCCESS in 5m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6ce0ef4aec494fe58b7db7115dd84fa6 : FAILURE in 6m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/42cec070b4d64a918882be2a2a45bc7c : FAILURE in 5m 53s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/4d341657fef8441caf3e4b89db6ce704 : SUCCESS in 7m 37s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ba52752f423a4411b6d5fe1b86623b75 : SUCCESS in 11m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/46db86bee92a4187b943763c890006aa : SUCCESS in 8m 05s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/05f611ce6f314e219b8a083bc46b49a1 : SUCCESS in 10m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1903d5625add46aca878192d6ffba930 : FAILURE in 8m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d990c68469cb43858ee6e4a1439c4bdc : SUCCESS in 14m 58s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/920c0ba530f64dc682ebf61510909a98 : SUCCESS in 18m 27s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5be6d37db0914eecab9e1773dca39365 : SUCCESS in 10m 00s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b4f358ee1f304116a62bc10d2899909e : SUCCESS in 5m 50s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7389397303a446039bfa561ff2206f6f : SUCCESS in 24m 35s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2f2cf3773455421f978d097f038a1446 : FAILURE in 15m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/628186b096514b61a5d73b060a4ffc6d : SUCCESS in 5m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b1946616ce5d4f3e8d4f6d973e7d2195 : FAILURE in 22m 35s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0dd36cd1ea764b37ab0beb11f2819103 : SUCCESS in 7m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e42c4ba88d754734afd071dd76c58f3d : POST_FAILURE in 8m 35s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6203d7cc82ae48ada9061ab6e5219c7a : SUCCESS in 12m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f3659d80b65463ea9ffffd3fbb1ed3b : SUCCESS in 49m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/11d3bc264cb74ecba2088ae580734df9 : SUCCESS in 1h 27m 37s","accounts_in_message":[],"_revision_number":1},{"id":"11448cca61de4e83047318dd13dafb51ea3103df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-06-20 06:31:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"321ac7f4ca594a18e431658d66df5d3d151eefa4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-20 06:58:51.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/bffaa74ec9b742cebd0b547b56f37568 : FAILURE in 10m 56s","accounts_in_message":[],"_revision_number":2},{"id":"8e200a34d5fa50d7ec53e91cdfb49a2585d9d52b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-20 07:40:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b088c4329f84ecf8b0782e02864aac3 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9db7166bd20445caa41cdb47a6d98d22 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b104be30e106451183695052ee0bfe49 : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ebfbe94692514f44bacb78d9fa1948dc : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89ad0a58b77f41429d98c8a4234ad750 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b5a76b08513a4c0c8ccd8ebfb04b400d : SUCCESS in 6m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f672d2dc242f4f7dbcb3b44f06f40dbc : SUCCESS in 4m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/376e887fe62642cc94e8c452ffd5a13a : FAILURE in 9m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/96a74e8671b84176bdb9638eb08d3382 : FAILURE in 6m 24s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f8e94c54a0a64902aab31ccf9bdcb718 : SUCCESS in 6m 59s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a20f38bf39d24f0b94afe0ecb060fe39 : SUCCESS in 13m 03s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7197349338e94f9abff7e7639de7b7ea : SUCCESS in 7m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e3c6826e55934dd1af15a409ae7921f7 : SUCCESS in 11m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4d5f9dc7fdbb4f2b9fb48e66bf49ea97 : FAILURE in 8m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/65d9f807668f4e0f8aa3aad0a0443878 : SUCCESS in 12m 33s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e0cf31222ae743f387493031394a3db8 : SUCCESS in 16m 13s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/39cfd76f416343b3a93311572a0aec37 : SUCCESS in 11m 07s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e457a4bf2eb6462da97d5a6c73fa2595 : SUCCESS in 7m 11s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/624ef408cb0e4deab806cf40b51569b9 : SUCCESS in 21m 19s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ad409db59dd14eda9b2fb11463eac94b : FAILURE in 14m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a71a6d70a0914079a8627c844e50654b : SUCCESS in 6m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/71e86e17406f455f989eafbf8a164558 : FAILURE in 26m 59s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1c4207a35a144a1a9beda466ed5ceba7 : SUCCESS in 8m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8fbd7ed22ca6452d8bcf557f0dffbcdd : SUCCESS in 8m 01s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0280e63ff0004c158f7e09a677ff270c : SUCCESS in 14m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ebe2b39b9744594b15cb0eeae354539 : SUCCESS in 46m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4c2a9341d6684de695d3879acfc864ec : FAILURE in 1h 06m 27s","accounts_in_message":[],"_revision_number":2},{"id":"c3cd58fa63095dbaf633390eb6eb137670144a14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-20 20:40:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4551777d8f600e236e5cd99c95264434aef8931a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-20 20:40:45.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":3},{"id":"9d4f41f7e8e028a6c7d4b40493a5a7d09858b2e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-20 20:59:07.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e888d5d91f7a404ebff4257bba5cb06b : FAILURE in 12m 57s","accounts_in_message":[],"_revision_number":3},{"id":"771f4fee5c8a60576c5e7ea24cfbf3afd9e28681","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-20 22:12:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0e823f7a4054c8881d27adc7288cbed : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8abd4dc0acca44aba567b384f42c785d : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e390b7956d2145038941d122f30bebdc : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/057ba651cc014fc6bb2d6200b3ca579d : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe8adf2d0c86430fbe4773d20f2d2719 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d2ffb2cc57314670beaf2a1017f219c6 : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3fdfb9a590a2430ab42a52c2a3e81173 : SUCCESS in 3m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/db5adc01f36c4ef0906ecf6d5c7fc920 : SUCCESS in 10m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fe4058881e9e4c1c82cdd6922df01ca0 : SUCCESS in 11m 40s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/8c3d00a770394980bf08029f348e11e7 : SUCCESS in 5m 51s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a737e172b2214031b4f05e0672db4585 : SUCCESS in 10m 14s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/cd3c91b991b744a29dcdf052b0aca8ae : SUCCESS in 5m 51s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7c597e24d5d04017854c40b75d2eeb5c : SUCCESS in 11m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/14d98aaa99d54d19a533ab69e747fee5 : SUCCESS in 8m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f77afc96088f4267a79f77ef26e7421a : SUCCESS in 11m 02s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b2c78b967e71402bbb9206baafa22bfc : SUCCESS in 14m 53s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/41437018eb9c47189a96c43808966960 : SUCCESS in 7m 58s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/20e73b92b52d4f80ae481549ebcc3ca6 : SUCCESS in 8m 00s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/652655195950486b850e52ca5833ea90 : SUCCESS in 19m 42s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3e4152ef80c84b71b7ca3debed101642 : SUCCESS in 13m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d515624a83264fc0b653225ca02212c4 : SUCCESS in 5m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/347125112c474564bffc8b5b3832274b : SUCCESS in 28m 42s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4bc695867dc64f57a74a5912b46aafa5 : SUCCESS in 5m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2e03b9b021094a2ca2fe11c4cfbb6813 : SUCCESS in 6m 43s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/30418d54ac0c439b8e10a24b784b5a14 : SUCCESS in 11m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/86be9f83537540188e60dd3e2a616503 : SUCCESS in 40m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0b203e2e45d24efa8977efa57a4e0dbc : SUCCESS in 1h 25m 59s","accounts_in_message":[],"_revision_number":3},{"id":"119f1e434ad38336aadf77a3744ef929edd6e194","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-20 22:21:52.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2b963f48ed0c152b83eaa9fed276ddb32520109c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-20 22:22:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d3d5b4c039ac2a764f538366c2696a1bea53434e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-20 23:29:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77e5372254ff4a4ea0d2ffef2ea43a1f : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cad522e7bc647799f5c699e9b9d7d52 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/538125007f644c23bbfd18cf0c5184bf : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df4bbb24ff4745fb800810b6b91015d8 : SUCCESS in 5m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/af58e72b4f18494f9c7a184f9538888f : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3559ec69ab224dd0a246ab8871b14b23 : SUCCESS in 4m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7c2230f720af44559cdf6d4f51cd179a : SUCCESS in 11m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5d6ea5e7ac35471eb2466c6bdce7f568 : SUCCESS in 17m 38s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8bfa097727374788a287cb102cb1628c : SUCCESS in 10m 19s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/082b95a56b9746ab994b9c46245e907f : SUCCESS in 5m 38s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/73a98906cff84fb1b5b567b8992dd69a : SUCCESS in 9m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a8ec0cbe68bf465ea6e99a11854c1a6d : SUCCESS in 8m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/407ad6a49a464623b59248ab8c952b70 : SUCCESS in 9m 53s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7e47c1148af74799bd6c282d81bbe336 : SUCCESS in 18m 40s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f3c7dfebaec6406e8f82a3af9e6457c3 : SUCCESS in 9m 25s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/943969cc5e1a46bc9876d78a59be14ad : SUCCESS in 4m 46s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/079a9789e3894ff4be8fb83865349094 : SUCCESS in 21m 58s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/83933d5aab3446e6923d8fece0b78411 : SUCCESS in 9m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/204259f879b7465784c27506fee1f0b8 : SUCCESS in 4m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f29739311fd342b0a3267e36d69d7096 : SUCCESS in 24m 36s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1e3c07a54787402a92e303ce3a50261b : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4355650ee0a14524a7455481d3e0d542 : SUCCESS in 7m 30s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6b6c0681afbb41c3a7629dae15a680f7 : SUCCESS in 9m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ad0c2361e9f94f30b15fb6db58966249 : SUCCESS in 42m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d9d85f9b8c2a4b76ac38f2a890542382 : SUCCESS in 1h 06m 08s","accounts_in_message":[],"_revision_number":3},{"id":"cf5ae2f9d4ac0b1e7690d6d01a7119bb8fc31558","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-20 23:29:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"6e4f3bc0a1ec2f3d52c1c0a0d0637f5599371944","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-20 23:30:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/756e935c03e247bca7bad0fd587101f8 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7772e841b75884792e995cb5125287f9d8ed3143","revisions":{"ade304ec117f41045dadf353d1eab5e99857ebe9":{"kind":"REWORK","_number":1,"created":"2021-06-18 06:34:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/93/796993/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/796993/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/796993/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/796993/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/796993/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/796993/1"}}},"commit":{"parents":[{"commit":"becd29d2c41a749e35eb7887f6ece3270d894130","subject":"Merge \"update constraint for ovsdbapp to new release 1.11.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/becd29d2c41a749e35eb7887f6ece3270d894130"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-18 06:34:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-18 06:34:37.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iecc516fc4be310f5bd85d46a00b596801c96562a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ade304ec117f41045dadf353d1eab5e99857ebe9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ade304ec117f41045dadf353d1eab5e99857ebe9"}]},"branch":"refs/heads/master"},"01924c5354161271e9c95ef4962ed34bfa5ec39e":{"kind":"REWORK","_number":2,"created":"2021-06-20 06:31:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/93/796993/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/796993/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/796993/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/796993/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/796993/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/796993/2"}}},"commit":{"parents":[{"commit":"960c75301338423c2b78c06742dca09ed30b07bd","subject":"Downgrade validation lib to 1.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/960c75301338423c2b78c06742dca09ed30b07bd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-20 06:31:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-20 06:31:57.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iecc516fc4be310f5bd85d46a00b596801c96562a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/01924c5354161271e9c95ef4962ed34bfa5ec39e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/01924c5354161271e9c95ef4962ed34bfa5ec39e"}]},"branch":"refs/heads/master"},"7772e841b75884792e995cb5125287f9d8ed3143":{"kind":"REWORK","_number":3,"created":"2021-06-20 20:40:38.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/93/796993/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/796993/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/796993/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/796993/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/796993/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/796993/3"}}},"commit":{"parents":[{"commit":"960c75301338423c2b78c06742dca09ed30b07bd","subject":"Downgrade validation lib to 1.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/960c75301338423c2b78c06742dca09ed30b07bd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-20 06:31:57.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-06-20 20:40:30.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iecc516fc4be310f5bd85d46a00b596801c96562a\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7772e841b75884792e995cb5125287f9d8ed3143"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7772e841b75884792e995cb5125287f9d8ed3143"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
