)]}'
{"id":"openstack%2Frequirements~797011","triplet_id":"openstack%2Frequirements~master~Ieb2cea16b6d9111986e4b8e377c560bcdf07cad9","project":"openstack/requirements","branch":"master","topic":"validations-libs","hashtags":[],"change_id":"Ieb2cea16b6d9111986e4b8e377c560bcdf07cad9","subject":"Downgrade validation lib to 1.1.0","status":"MERGED","created":"2021-06-18 09:21:53.000000000","updated":"2021-06-19 00:25:49.000000000","submitted":"2021-06-19 00:24:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"797011-validations-libs","meta_rev_id":"d361d19f48a17f7adb434c9dc1892f3d3bfc4371","_number":797011,"virtual_id_number":797011,"owner":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-19 00:24:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-18 18:52:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-06-18 18:52:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-18 10:11:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-18 12:33:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-18 18:52:12.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"5278ea128be42554c58c000cab3d0aa4bbe39d6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2021-06-18 09:21:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37a65b348dc5de14a68197d595ffa0f4a3e8e00e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2021-06-18 09:41:15.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1a985a111f83f14d8b26c141326a78da85aab3cc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2021-06-18 09:44:55.000000000","message":"Topic validation-lib-regression removed","accounts_in_message":[],"_revision_number":2},{"id":"d3dfa73be27404d9eaf515cc01587b16718d9531","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2021-06-18 09:45:03.000000000","message":"Topic set to validations-libs","accounts_in_message":[],"_revision_number":2},{"id":"3e146f96730ff1ffd528889aa91889f08690f542","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 10:11:28.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/97053207a6494afb815c202e93a78168 : FAILURE in 24m 20s","accounts_in_message":[],"_revision_number":2},{"id":"7bc178af8c33064c8e1041e2558dc662bdd73f01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 12:33:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a06462755654688a967f5361acc366f : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b7b857d4115475a87531ee21c313ab8 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04ae7958aea940099f6f4c76c98eff09 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db7941e13fd14c2681cc3947bcaa2fc5 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec7ecee35cc0432898532b2003b4f07b : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c3ca50a784174534b3538f37d2676c70 : SUCCESS in 4m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1e51179c212c43f0acac0174a02f9eb2 : SUCCESS in 4m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/09d19a8c53a24ee781a5204cf35644c5 : SUCCESS in 13m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4ede75c8c9bd42f0800646e40039d55f : SUCCESS in 11m 04s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b85bd87b7ce44b24b12c1cbf94592ef7 : SUCCESS in 6m 44s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/417ba07527634896929bd8b014da3c4c : SUCCESS in 11m 42s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5f4ec805cfc34ac5863aaf31d793d5d1 : SUCCESS in 7m 20s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8b2b001578d44fe3bd66d843c4e11813 : SUCCESS in 10m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/801c94816d6b4c7fbb785da6a8488a08 : SUCCESS in 9m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/07f75f4882e14143ba80472dbeef02ca : LOST in 15m 40s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0b5738bc86c945d19cef9396bbbcecf0 : SUCCESS in 16m 52s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/bd3e3598ec894314a4c92027173def14 : SUCCESS in 13m 42s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b4e869db4e414d7eb0c9fc49f6e5caf2 : SUCCESS in 5m 44s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3a4bf29aabeb49fe9786ea47b3aa0d46 : SUCCESS in 29m 15s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/01b9e5f31c8e4b938c743443dbb8e46e : SUCCESS in 14m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/594c6661b75e43d2a4dc69e57f89acf5 : SUCCESS in 4m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/04db2e82442a49b98fc99293e7bdc838 : SUCCESS in 21m 42s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f44d100a5395492f9afe3ad8f652a17f : SUCCESS in 5m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/09a6cdf7d71e4b74a96f472bdeb37267 : SUCCESS in 7m 39s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a21510fe36954343ae4a8f35bc7bba7b : SUCCESS in 12m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af8a33377b3e4024970d081f2e3f20e4 : LOST in 10m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d38a86e02eb949d1a11d653d65d44259 : FAILURE in 1h 04m 26s","accounts_in_message":[],"_revision_number":2},{"id":"1e210d2bbe288883c08ffc52fe82beb0eb5e1cf4","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2021-06-18 14:06:30.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e24b3a987494a20a6d0487f0c63171b28d9f3dd4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 14:26:53.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/35337870556d4dedbec127aba2f5b83b : FAILURE in 18m 24s","accounts_in_message":[],"_revision_number":2},{"id":"401db03ce5624956215d3e5f25da724a662240f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 17:09:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc65d5310ff64232a923934b5e8326e4 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57f39c83d7434dc48856b90105c954e5 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f673eb49bea24d74a835d383f33be3a3 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e964a20ff774d9792bfa183e2c47abc : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/379175a166a4434cac28715b14600f42 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/807c34141e664247bff9d489eaee5ca4 : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9c4288e07c654b77847ddd463ef1bd5f : SUCCESS in 4m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d54909c336824009af8d691bb9e8ef33 : SUCCESS in 14m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0136f2c46c5849b5932d8b84a08c70c2 : SUCCESS in 12m 00s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/8fb718297598429e9adfb65250609e57 : SUCCESS in 7m 14s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/261f102f260342ea992813655c69fe02 : SUCCESS in 10m 44s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/51a95f3072cb4f36b6c9f1b1e79b85d9 : SUCCESS in 6m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8225ebeb6031491081a272b23f31d726 : SUCCESS in 12m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9bdadd5ad16347ed832c7bd7d4fd5714 : SUCCESS in 10m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d2ca62edfb0a4594b06f77110039879e : SUCCESS in 17m 52s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d3cb41c47048472694e7cb286f78f46d : SUCCESS in 14m 42s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2ea7fb4483db4079b58b654eb74aaaef : SUCCESS in 11m 27s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a181d126928e4dd8997a6c2a96dbcaf8 : SUCCESS in 6m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/5c8607e08b124ee49ab0242014f03a8c : SUCCESS in 24m 05s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/57da0a1be37c4723bc61623e4e615a69 : SUCCESS in 16m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c43469eb95db460ca99065e3850a6834 : SUCCESS in 4m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ff93a1c6a4e7473291252550abc59842 : SUCCESS in 20m 57s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a605ab6cc3544150b3aa49ff484cb066 : SUCCESS in 6m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dbc46114384c42f88d2796dd3f0af4e9 : SUCCESS in 7m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/01555669349147fda1997cab316b0a01 : SUCCESS in 13m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4af4cb4600954f53bd321c388d43c8fa : SUCCESS in 54m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/109fa0368bd6493bbdd7f2762ef44160 : SUCCESS in 1h 36m 16s","accounts_in_message":[],"_revision_number":2},{"id":"8fb048876b95e6c3fbc2cf194cfed2e28bde6790","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-18 18:52:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"aaf0ee393abe7442f02cfc8fb0a9f34207e534d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 18:52:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d6e724848936e147b1a6092943915fe156230002","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-19 00:24:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4dfbb11f58f482ab69518d8378dc489 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0edc6aaff9184b02b097fe0d6c632fed : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c0a9682e4c140cabe84495f3142ea6a : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c25d9c2f6d8045f6bc4e47e97b6e0dfd : SUCCESS in 5m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/344cd2fff019457b88ebdd0509664ea8 : SUCCESS in 5m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5d281e19e4de4180a8c0384e4162be76 : SUCCESS in 5m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/34bdd2d6134342df89bec7c048f81569 : SUCCESS in 11m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f5437b6bb72d4604b931daf0fd191def : SUCCESS in 11m 02s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c7422b1cb9ec4b87aab3582b9c76c632 : SUCCESS in 11m 30s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/56afcbc8b31e414487ef37ea1fceacf8 : SUCCESS in 6m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6e531aaf4a6144c89a87297ac1b2b1ee : SUCCESS in 10m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cd3d25c9997b4fefa9466600cafebd0e : SUCCESS in 9m 18s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fa9e0c4ec6894f5a8f5e4297a0498bcd : SUCCESS in 10m 51s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/cc5ac7ab4c4f4561892e026fc656ca16 : SUCCESS in 27m 25s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/fafc6e9e266d4cf99645f02a143c463d : SUCCESS in 11m 41s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/44314e0695e94220a7f2aec48f4110f6 : SUCCESS in 9m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/0792b20238404b58b1660d8f49297384 : SUCCESS in 22m 04s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/bbc556bf8cbd4ab586019ab35f7c9853 : SUCCESS in 17m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6771475dd6764b808f555fea1c685685 : SUCCESS in 4m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/402dbfaec9cc4e8d97b8e906f6fe06d7 : SUCCESS in 19m 21s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/59d0e3555f0c402082dcc1075f6f02ee : SUCCESS in 8m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2836bf650b5248c1811d60e17fa5bbcd : SUCCESS in 7m 41s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/18574eef6414437498c504ae2db7df5d : SUCCESS in 24m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/521daa48048548ccbeddf2ddc3991887 : SUCCESS in 43m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/07e89389eb734945ac5c8a80a0a5f999 : SUCCESS in 1h 36m 22s","accounts_in_message":[],"_revision_number":2},{"id":"68745250d14657793b880752d4add836b77d0808","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-19 00:24:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d361d19f48a17f7adb434c9dc1892f3d3bfc4371","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-19 00:25:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/339bbc5af2704cc685605ea0c3cc44d6 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"960c75301338423c2b78c06742dca09ed30b07bd","revisions":{"9771fa66f8e5bf72d525fb5310c42700f1d3fa66":{"kind":"REWORK","_number":1,"created":"2021-06-18 09:21:53.000000000","uploader":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"ref":"refs/changes/11/797011/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/797011/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/797011/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/797011/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/797011/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/797011/1"}}},"commit":{"parents":[{"commit":"becd29d2c41a749e35eb7887f6ece3270d894130","subject":"Merge \"update constraint for ovsdbapp to new release 1.11.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/becd29d2c41a749e35eb7887f6ece3270d894130"}]}],"author":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2021-06-18 09:19:01.000000000","tz":120},"committer":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2021-06-18 09:19:04.000000000","tz":120},"subject":"Downgrade validation lib to 1.1.0","message":"Downgrade validation lib to 1.1.0\n\nVersion 1.2.0 has a regression performing expansion\nof $HOME variable [1]\n\nThis proposes to pin previeus version while waiting\nfor a fix [2] ends up in follwing release\n\n[1] https://opendev.org/openstack/validations-libs/src/tag/1.2.0/validations_libs/constants.py#L25\n[2] https://review.opendev.org/c/openstack/validations-libs/+/797008\n\nChange-Id: Ieb2cea16b6d9111986e4b8e377c560bcdf07cad9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9771fa66f8e5bf72d525fb5310c42700f1d3fa66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9771fa66f8e5bf72d525fb5310c42700f1d3fa66"}]},"branch":"refs/heads/master"},"960c75301338423c2b78c06742dca09ed30b07bd":{"kind":"REWORK","_number":2,"created":"2021-06-18 09:41:15.000000000","uploader":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"ref":"refs/changes/11/797011/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/797011/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/797011/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/797011/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/797011/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/797011/2"}}},"commit":{"parents":[{"commit":"becd29d2c41a749e35eb7887f6ece3270d894130","subject":"Merge \"update constraint for ovsdbapp to new release 1.11.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/becd29d2c41a749e35eb7887f6ece3270d894130"}]}],"author":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2021-06-18 09:19:01.000000000","tz":120},"committer":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2021-06-18 09:40:28.000000000","tz":0},"subject":"Downgrade validation lib to 1.1.0","message":"Downgrade validation lib to 1.1.0\n\nVersion 1.2.0 has a regression performing expansion\nof $HOME variable [1]\n\nThis proposes to pin previeus version while waiting\nfor a fix [2] ends up in follwing release\n\n[1] https://opendev.org/openstack/validations-libs/src/tag/1.2.0/validations_libs/constants.py#L25\n[2] https://review.opendev.org/c/openstack/validations-libs/+/797008\n\nChange-Id: Ieb2cea16b6d9111986e4b8e377c560bcdf07cad9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/960c75301338423c2b78c06742dca09ed30b07bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/960c75301338423c2b78c06742dca09ed30b07bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
