)]}'
{"id":"openstack%2Frequirements~797048","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I208b00629493953a502f8b95756820ae0d9a9cb6","project":"openstack/requirements","branch":"stable/wallaby","topic":"add_ruamel_yaml","hashtags":[],"change_id":"I208b00629493953a502f8b95756820ae0d9a9cb6","subject":"Add ruamel.yaml to global-requirements.txt","status":"MERGED","created":"2021-06-18 10:20:02.000000000","updated":"2021-06-25 03:29:45.000000000","submitted":"2021-06-25 03:29:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"797048-add_ruamel_yaml","meta_rev_id":"e709497db35dd05f7de8b96e02a805012907e88d","_number":797048,"virtual_id_number":797048,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-25 03:29:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-18 19:04:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2021-06-24 13:45:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-06-24 21:09:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-18 10:23:40.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2021-06-18 10:23:40.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2021-06-18 10:23:40.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2021-06-18 10:23:40.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"state":"REVIEWER"},{"updated":"2021-06-18 10:23:40.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"state":"REVIEWER"},{"updated":"2021-06-18 10:23:40.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"state":"REVIEWER"},{"updated":"2021-06-18 10:24:08.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-06-18 10:35:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-18 13:26:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cac965e46c7131cf6eee226e8446620353e27b60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-06-18 10:20:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28bac0e48529da4b813449eb3af38d21e9304f02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 10:35:16.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8d541351e26c4445bf802ad70ebe5377 : FAILURE in 10m 51s","accounts_in_message":[],"_revision_number":1},{"id":"82d55c5a4153c05615d84b98dc580ca0e5c15cc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 13:26:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2229a0048a004f3eadf1cdf33cd54a9f : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80618346f0744e84be36098a5f808eb7 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5aa4008a9a9d4322b7df02d3b7af4ffd : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e0ef40726b34164ad860f01c5a38d36 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e9b914e1fe6148479b0f46c511737530 : SUCCESS in 7m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0680c501a64c4566a05a88d1d7837c7b : SUCCESS in 5m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5a5cd66bbcd84e5e9aaee3d281614537 : SUCCESS in 10m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/94e8eae3f1c94b98ae3f3c00012cbcd2 : SUCCESS in 10m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/579dbd1c58734692812b2724e4a4b20c : SUCCESS in 10m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/382000ed832b48129caef24e00b697b6 : SUCCESS in 6m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4ad3ce487c6145abbed80de85a67f481 : SUCCESS in 10m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0f2480c3fa044655aa13654ba5747a9c : SUCCESS in 7m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8ede31601efd4fd191f2963f89748c4b : SUCCESS in 10m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a5e49709708e483d88cc5e01a468a7fc : SUCCESS in 14m 37s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/37a82507b7754928b6161948dc66cb3b : SUCCESS in 4m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/70a1c0cec1224af5a57f314130d4ac76 : SUCCESS in 29m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c5c75f8900b7454a80c70263ae7e4562 : SUCCESS in 13m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bf5067423bb148a187b9343603a747aa : SUCCESS in 5m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c975e957c97c4adea0261de40a12e375 : SUCCESS in 18m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e3047bf4975a4a928aceb249f7e7e394 : SUCCESS in 6m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/06fe8179e0374ae3b5c455fd1cc7ec63 : SUCCESS in 8m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/87c400cc6a314499bbef3d0ed465478c : SUCCESS in 12m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/39cefe2ee3814eedbf7f0842b0bdb146 : SUCCESS in 59m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b71f561f1e04808aeb2d6f975852ae3 : SUCCESS in 1h 26m 36s","accounts_in_message":[],"_revision_number":1},{"id":"12a38864fa316ea822c6e5bfef16f726a8448b22","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-18 18:53:19.000000000","message":"Patch Set 1:\n\nhmm, stable branch gets new dep because docs generation spans branches, not normally allowed but I think is understandable.","accounts_in_message":[],"_revision_number":1},{"id":"da156f84378a0e6291f3276acf13eba3429f77e6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-18 19:04:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9219ee48914b14c74b72f135ffb6864e86618f73","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2021-06-24 13:45:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bc1da37b569f3bed23ef2f739c0668c0f4808b5c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-24 21:09:09.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ea81a88d878a42104ee1c75fe8f2fba4a23b1be4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 21:29:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9541ad402d72cad80beb62ed3f40c02c37787ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 03:29:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5d56d7b200d4121b3b775e10d2d5bf3 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e71dab1cc884212b8ede321c07a97e5 : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aed390d2a8fc4a9b921d14dc3bc77bf6 : SUCCESS in 6m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/101b9839582e4da68d8089283526210d : SUCCESS in 7m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/23dd93555aa84be4a888f285bd198795 : SUCCESS in 7m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0949e1467a8b4051b01db1a806a9836c : SUCCESS in 11m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6b1f391b28ce45c7a5392293b5595119 : SUCCESS in 14m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4bfc30dcfa6f4809805be0f87d3f6eba : SUCCESS in 13m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4cde164282ea4b0d8f11791bffb1898d : SUCCESS in 9m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2b5d53c54cae48c4a7f2c73ae4119e88 : SUCCESS in 11m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2dc0ac919f5940089b9156ebed4732d8 : SUCCESS in 11m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a4e40f3efec04fb2961d38307aa740b6 : SUCCESS in 14m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/05457f7bb32f4e0dac26f91cf0a87d5f : SUCCESS in 13m 37s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/82055e922fdc4fb28e3003a1b5efab77 : SUCCESS in 8m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/998bde44e71b4d438426b17bbace492a : SUCCESS in 28m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fd2d527b2a27407b873ae668acd29d7d : SUCCESS in 16m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0d29b4aa5d4e4bcbbff360970479d13b : SUCCESS in 6m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/02e120240cce4e94898609d8bd088fc8 : SUCCESS in 23m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/15b926b5362c4f7b9edaac056f7e26f5 : SUCCESS in 7m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fc658edd6f8746e1ba6f3775e6c329b1 : SUCCESS in 10m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9104c35f07374005abb382a02fc1f399 : SUCCESS in 15m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e61c8f43c96a4fe38976031365431d80 : SUCCESS in 43m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dafee4eeacb248aca114ea50b2eb1161 : SUCCESS in 1h 28m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e709497db35dd05f7de8b96e02a805012907e88d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 03:29:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"148563d374d3cad33c286f28a10e4ba330d154b5","revisions":{"148563d374d3cad33c286f28a10e4ba330d154b5":{"kind":"REWORK","_number":1,"created":"2021-06-18 10:20:02.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/48/797048/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/797048/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/797048/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/797048/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/797048/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/797048/1"}}},"commit":{"parents":[{"commit":"aee7a4b34ee91065581ae20ba81d30920bb18769","subject":"update constraint for sushy to new release 3.7.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aee7a4b34ee91065581ae20ba81d30920bb18769"}]}],"author":{"name":"Gael Chamoulaud (Strider)","email":"gchamoul@redhat.com","date":"2021-06-08 10:04:56.000000000","tz":120},"committer":{"name":"Gael Chamoulaud (Strider)","email":"gchamoul@redhat.com","date":"2021-06-18 10:19:23.000000000","tz":120},"subject":"Add ruamel.yaml to global-requirements.txt","message":"Add ruamel.yaml to global-requirements.txt\n\nThis Python library is now a requirement for building\ndocumentation (ansible-autodoc) for the following project:\n- openstack/tripleo-validations\n- openstack/validations-common\n\nruamel.yaml is a YAML 1.2 loader/dumper package for Python.\n\nThe library is actively maintained, with almost a weekly release. It is\ncompatible with Python 2 and 3 and licensed under the MIT.\n\nhttps://pypi.org/project/ruamel.yaml/\n\nSigned-off-by: Gael Chamoulaud (Strider) \u003cgchamoul@redhat.com\u003e\nChange-Id: I208b00629493953a502f8b95756820ae0d9a9cb6\n(cherry picked from commit 5930f2fbd44667943b9ea20fcaf995f25dde4954)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/148563d374d3cad33c286f28a10e4ba330d154b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/148563d374d3cad33c286f28a10e4ba330d154b5"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
