)]}'
{"id":"openstack%2Frequirements~797227","triplet_id":"openstack%2Frequirements~master~I4afcf08726536d51c20e7d9fa4cf1b5b0057efd2","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I4afcf08726536d51c20e7d9fa4cf1b5b0057efd2","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-06-21 06:32:33.000000000","updated":"2021-06-21 23:47:59.000000000","submitted":"2021-06-21 23:46:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"797227-openstack/requirements/constraints/noclob","meta_rev_id":"43155a5bb014de07c2a7ac785e85ad45866f5826","_number":797227,"virtual_id_number":797227,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-21 23:46:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-21 17:59:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-06-21 17:59:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-21 06:47:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-21 08:14:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-21 14:59:06.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"2ece3216949a609b3a05b859b8d97e33a14f0ade","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-06-21 06:32:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f03c617d766f4119589471632fcc70950fd223a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 06:47:41.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e1ddc5271e834f2893d29a13b418429c : FAILURE in 12m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e54434262f616a5fc5f6173180a6bf03c3196706","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 08:14:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1a73ff4834240dcb9d97cd7fd471ad0 : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c40b9236e0554c01a36b71bd3df5d143 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a51d588468d4955a0508bba5e77f007 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00c797bc81c64d72bca9a945dcf9593a : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2c4605aa26b4489b45971ca8c73e6bc : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a009325ba9f348eda8956ed79a4caf78 : SUCCESS in 5m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c71e70a8766a476ba5336baf1a81a430 : SUCCESS in 7m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/46347d9cc49049c3b077a019e14e9c81 : FAILURE in 8m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8fa72896648a4326b68193b8241b379b : FAILURE in 7m 39s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/71e2eb8d20754c5199a30badf0d5019f : SUCCESS in 7m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5af0ac339c164ccdb46c7f7057120a29 : SUCCESS in 11m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/13c15a564ce84b2ab29db254ee81df60 : SUCCESS in 7m 57s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b07415ab1fda4e2ab123cfa2c26b925c : SUCCESS in 10m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0628dbe728e14c05bb49961ece587994 : FAILURE in 7m 49s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a67a70e156cb4e3e8ce8f0bd5093257e : SUCCESS in 16m 08s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/9281dbde0f6e481b87323d06d1ef5ca7 : SUCCESS in 15m 47s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a68bc5325119458e908b502bd8fdc4dc : SUCCESS in 11m 47s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2adb24c8ab7d404aab440baf1f26c780 : SUCCESS in 7m 19s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4ce770345fe74ff4a0a60653468984d8 : SUCCESS in 22m 11s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5bb9a3d9caf64e6abccf85aa7f622d96 : FAILURE in 13m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/698a09c7eecd4f38963f8056d922b665 : SUCCESS in 5m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8925cc5f3faf43fbb7ae1775d2042069 : FAILURE in 20m 50s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9c350cbe269e4ef3807e36b8ef0684e4 : SUCCESS in 6m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f081558cadb446638745343f00e00ba4 : SUCCESS in 7m 01s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6b290f663aa94ac2858d1facc3d01b80 : SUCCESS in 12m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d5b3b54bf5c34ac0b1b3dc483f0bd7b6 : SUCCESS in 48m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e4ae29e0759643c78edd39d7ca9de4d4 : SUCCESS in 1h 25m 22s","accounts_in_message":[],"_revision_number":1},{"id":"a37976ab7a401f9b2d56a77697cf1284312a728c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-21 14:59:05.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":1},{"id":"00f2114c903dc000326faf19eed7a227d6df0ad7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-21 14:59:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c5b3ee7375a2d318fdc361522f8c61e57387e1f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 15:18:28.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/cf267a0e018a47c08adcfe8cc1c42fa0 : FAILURE in 17m 15s","accounts_in_message":[],"_revision_number":2},{"id":"a690aa5f674472e9f39c52c2f0006480e2391aee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 16:14:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15bec794c7a14b9ca2f765c5a842ac47 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5ede8daaf004cb2b7cd36087a85e5f7 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7af360cde8b41fc9763e18870adca88 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71a29ea5397749cabb780e94e02d7c07 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/caf45e66fe6c4b5e9fd87c31b4054ed2 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/62fd746e838044dfaa8072703f025258 : SUCCESS in 4m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9a6340b56c9b4be396209c847f863218 : SUCCESS in 5m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/db921a3092c3446cbd658f40845964ab : SUCCESS in 11m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/be259b37e188475ab583f7ca7776255b : SUCCESS in 11m 53s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b906e17082a042d7b004b142080a4cbc : SUCCESS in 5m 53s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e6cfefbaba214ff4a5d374348aad9dca : SUCCESS in 10m 32s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ee09ddf7d9f24f51a621e43f9120c5b5 : SUCCESS in 5m 59s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e67cc67e6a6d47cc8515a1976e32cf4d : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/18c5c67733ef497eba1c1a38ecabd369 : SUCCESS in 12m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4842ab81eea348ca8af2dcb653cae8f2 : SUCCESS in 13m 19s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/23925e6a2076471384d50ba03255c478 : SUCCESS in 14m 58s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e44c0236b58346e69088a33b04569644 : SUCCESS in 10m 14s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/964768cb8a944f67bec3a5847513a615 : SUCCESS in 5m 57s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/0e3ddcd322fc40438f9a9ddd895313c6 : SUCCESS in 22m 02s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e2e2e8ed9a75410ab03816f3aab9db9d : SUCCESS in 13m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f70b442bd68d4eb6bd9cb32d2015bead : SUCCESS in 5m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4ce19667f86e4afda10309c6bd7181c5 : SUCCESS in 28m 38s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a535cdf825b144d4bca6325f68146d95 : SUCCESS in 6m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8a0b5ad3bb1d4c6e8f31f323c4afde3e : SUCCESS in 7m 00s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/eb7ab8b0641448729d431ee242b935a0 : SUCCESS in 11m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51868ff3dc9248a18fa42a5bbbe641af : SUCCESS in 1h 00m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b49d860f18b4a01aaae454ec176168b : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":2},{"id":"790eb9989b1962fc3de5a7c3cf08ac3f85febb39","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-21 17:59:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cf99373d9db937aafe9856104afdfd0a7ff3f442","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 18:01:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d7b1177c1628344d77432a101df932b1e3331f5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 23:46:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36ef348b8ae84da28c938b8d9bd84c6a : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c94268de96754ee9a806b4412c34b88c : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/628465cccc5540bb94484cf2b4a2373e : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da41c14d85484a83b2acad8d6d52c7fb : SUCCESS in 5m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/015a523a98ca40879f9c867d54581411 : SUCCESS in 5m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3320f2e76d9b477f8c2705983d17198d : SUCCESS in 4m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f11a1c3a8d2249b6a71b59272858eef2 : SUCCESS in 11m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/34fe92faa5fc47e6a4d17692095d66f8 : SUCCESS in 10m 31s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7305313af35b46849726d42ce94b3719 : SUCCESS in 13m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/85939c981674467c9090f90cb6403095 : SUCCESS in 6m 47s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/87e71bfe4c9a443192525fc27ec73c50 : SUCCESS in 9m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a99f31faf0b143049220d3ded6743555 : SUCCESS in 8m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ce4d4f8c500c4907a7ddfa3674694191 : SUCCESS in 18m 27s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6e2446bc07444d048b06e2b477a4456c : SUCCESS in 13m 20s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/cc9a4aa19b224800bf9aec9e82fb2066 : SUCCESS in 10m 11s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/023ab6fe8f2d43edaff5f2a7536ce9bf : SUCCESS in 6m 32s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/183bcb92a8404eb8a300124a611d760b : SUCCESS in 21m 39s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/deeca1557b38499ab0d550b95b9320cf : SUCCESS in 12m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9c4865e533a34d8e9681a2c779a9ed70 : SUCCESS in 4m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cd51f4756c3e4ddeb34db713e30892fa : SUCCESS in 25m 26s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e1da496e966e4de48313350c371948f0 : SUCCESS in 6m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c3d40e7facde45869c215de9f0da5870 : SUCCESS in 15m 05s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/42d2150bfb5241ea859f5a850caf0b04 : SUCCESS in 14m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2961a5b8a27a47dd9073af00c1003dbe : SUCCESS in 45m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97fe1b18c81b433c9c23e6ff4eca0759 : SUCCESS in 1h 14m 06s","accounts_in_message":[],"_revision_number":2},{"id":"a5d34d751f16c9f62c2ff0f55eeab6db654e00fd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 23:46:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"43155a5bb014de07c2a7ac785e85ad45866f5826","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 23:47:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d2056acf8224856aad263d1ff43f20e : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"269a7bc5ffba8a0830c20fca6a31a45612c9c3f6","revisions":{"6558d3db639b8df2f7abf0d988ebd62b69eb9f28":{"kind":"REWORK","_number":1,"created":"2021-06-21 06:32:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/797227/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/797227/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/797227/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/797227/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/797227/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/797227/1"}}},"commit":{"parents":[{"commit":"a2e4ea54f3c06c93c853dab93d6cfd7e1a75d4da","subject":"Merge \"Use py3 as the default runtime for tox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a2e4ea54f3c06c93c853dab93d6cfd7e1a75d4da"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-21 06:32:30.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-21 06:32:30.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I4afcf08726536d51c20e7d9fa4cf1b5b0057efd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6558d3db639b8df2f7abf0d988ebd62b69eb9f28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6558d3db639b8df2f7abf0d988ebd62b69eb9f28"}]},"branch":"refs/heads/master"},"269a7bc5ffba8a0830c20fca6a31a45612c9c3f6":{"kind":"REWORK","_number":2,"created":"2021-06-21 14:59:06.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/27/797227/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/797227/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/797227/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/797227/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/797227/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/797227/2"}}},"commit":{"parents":[{"commit":"a2e4ea54f3c06c93c853dab93d6cfd7e1a75d4da","subject":"Merge \"Use py3 as the default runtime for tox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a2e4ea54f3c06c93c853dab93d6cfd7e1a75d4da"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-21 06:32:30.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-06-21 14:58:52.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I4afcf08726536d51c20e7d9fa4cf1b5b0057efd2\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/269a7bc5ffba8a0830c20fca6a31a45612c9c3f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/269a7bc5ffba8a0830c20fca6a31a45612c9c3f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
