)]}'
{"id":"openstack%2Frequirements~797615","triplet_id":"openstack%2Frequirements~stable%2Fstein~I1210b900f1c3f4608eeb6d289dc0a9600b6f47d0","project":"openstack/requirements","branch":"stable/stein","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-07-19 08:56:39.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I1210b900f1c3f4608eeb6d289dc0a9600b6f47d0","subject":"DNM: requirements CI test","status":"ABANDONED","created":"2021-06-23 09:53:52.000000000","updated":"2022-07-19 08:56:39.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5cde5011ca0ced91b5dce588e5591dccc9230bbf","_number":797615,"virtual_id_number":797615,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-06-23 11:20:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-23 11:20:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-18 23:29:14.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"}],"messages":[{"id":"121eef42d2c450d5d476866e563519a83b2555ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-23 09:53:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf51be81a5cf0ed6fc2eb9190423a4f277a94e36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 11:20:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53b5037439934b7f90b883d1e20dcac7 : FAILURE in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b695e9d0fd3b4e33ac5622f60a210702 : SUCCESS in 4m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8362c12de2e043e38acedd576a6a6f99 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dab5062ecb6c4f07b9fd84a460bffb52 : SUCCESS in 5m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2a5912f83c5943158aa8ebfca5107ad0 : SUCCESS in 4m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4d06298a3d9d4671885b66327843fe62 : SUCCESS in 55m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b78ae6d142a84c7e9ebf6ec2948a35fd : SUCCESS in 1h 22m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/487277225fa8438699022e61a640aa34 : SUCCESS in 58m 24s\n\nWarning:\n  Comments left for invalid file openstack_requirements/tests/test_update.py\n  Comments left for invalid file openstack_requirements/tests/test_check_constraints.py","accounts_in_message":[],"_revision_number":1},{"id":"ac2fe4943e77a5a507a7a86c29cc5cbc572907fa","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-07-18 23:29:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5cde5011ca0ced91b5dce588e5591dccc9230bbf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-19 08:56:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6e34c4ffde3aafb6670d3d47dd8408fbd22e5e99","revisions":{"6e34c4ffde3aafb6670d3d47dd8408fbd22e5e99":{"kind":"REWORK","_number":1,"created":"2021-06-23 09:53:52.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/15/797615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/797615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/797615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/797615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/797615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/797615/1"}}},"commit":{"parents":[{"commit":"45797e0e4336a65af6d8a7d31dde507b08dd6570","subject":"update constraint for neutron to new release 14.4.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45797e0e4336a65af6d8a7d31dde507b08dd6570"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-06-23 09:53:08.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-06-23 09:53:32.000000000","tz":120},"subject":"DNM: requirements CI test","message":"DNM: requirements CI test\n\nChange-Id: I1210b900f1c3f4608eeb6d289dc0a9600b6f47d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6e34c4ffde3aafb6670d3d47dd8408fbd22e5e99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6e34c4ffde3aafb6670d3d47dd8408fbd22e5e99"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
