)]}'
{"id":"openstack%2Frequirements~797654","triplet_id":"openstack%2Frequirements~master~Ib727242c1aba8597389259d8dbacba7194f9f0d6","project":"openstack/requirements","branch":"master","topic":"fix/generate","hashtags":[],"change_id":"Ib727242c1aba8597389259d8dbacba7194f9f0d6","subject":"Make tox -e generate easier to use","status":"MERGED","created":"2021-06-23 13:54:39.000000000","updated":"2021-06-25 02:41:37.000000000","submitted":"2021-06-25 02:39:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"797654-fix/generate","meta_rev_id":"c55fe34924cda676a3f42eccf46ef1624c5d248b","_number":797654,"virtual_id_number":797654,"owner":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-25 02:39:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-24 13:00:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-06-24 13:20:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2021-06-24 13:20:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-23 13:56:30.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-06-23 16:18:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-23 16:41:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"a46ed4e2dabaad0d4ae88830cc38763b78c11150","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-06-23 13:54:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ade492e6dfed2acd878df18107e75fb1ccfb24be","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-06-23 13:56:30.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"a62a1f16f9f88c6bb6144180aa3652967716da64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 16:18:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b54f2ecca92d4521b499abb88cc5ed7b : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a68ecd349c5e4f1fa1d7a4f867ae6265 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99f7e48d825442f1a364e412ff19b676 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6267c655fba4a25bf602481d35ae9e8 : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54e20f22408743db9ef5b214a2cec0f6 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/47d34f5cedc048fcb80f3f96891be7de : SUCCESS in 5m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6d2ee1e60fb84da794e6df47e4436c2f : SUCCESS in 4m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c15eec7f678466da3e78db43b42b256 : SUCCESS in 52m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/186abbe173bc41dca44cea6d5c84e3fe : SUCCESS in 1h 27m 49s","accounts_in_message":[],"_revision_number":1},{"id":"b30f6b5ca2394a28184df8373584ddedf5d253f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-06-23 16:34:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1e800763c9b383b88ee6f75e681fb9c671372cb1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-23 16:41:22.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4b49cd9c6585fb2e8f74b3e7c50067f2e2ee628d","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-06-23 17:37:28.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"18e2c756e8e622b0d944bcfe258970392311fcf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 19:18:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edea7ce24b304038b43736517b91e40d : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b4aff3664534d40b66fb371ad8397ed : SUCCESS in 8m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06fe62b56ccb4e1ab0c6c0f060a48ed2 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/113a4ba51f964c6bb2b13e4f9f8d5db1 : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c652f527407943d5971fb6836680dcdb : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3096a08311b64e4681e3d40dbede158f : SUCCESS in 6m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1c4d1e3801434900881186edb9e526a7 : SUCCESS in 4m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a30810654779426b99f4f4c90510d764 : SUCCESS in 50m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b8e687c91bb4c988da41bf1f44bd80d : SUCCESS in 1h 30m 24s","accounts_in_message":[],"_revision_number":2},{"id":"4fad1a1940d094a1e450b81668384772a22ee84b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-23 19:44:39.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f5c87722c39ea481be546a729886bea57f85d403","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-06-24 09:29:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"19dcd5cde16d33a53bd6508231a9da99de7e32b7","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-06-24 09:34:10.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9c5fba4d5ec3daa36455762419648e2bdf3e099f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 11:01:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e831fc74cf81418f81d7882c7526ca28 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f57e94fc1b5b4d0ebb45ab7df913a5ad : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe2ee1cfed2d40aa96f9f3a1930eb90e : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05ab319282bd4a2886acb1dc18d66323 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3480a29f18a457f836ce9cdb2006faf : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac980b37f41f478e8fa0bc4254b07816 : SUCCESS in 5m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e38708bb05e74d0bbb002041b89883b4 : SUCCESS in 5m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/470585c09bab4fd2a5830a8bb4a0c38e : SUCCESS in 48m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/30c0850c95be4daca65fb6449b66e852 : SUCCESS in 1h 21m 37s","accounts_in_message":[],"_revision_number":3},{"id":"6a704261b3287404849ed874ba034b1126c19f19","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-24 13:00:18.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2b7d9d094407b24db33d5ab9d358a31cf57e4145","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-06-24 13:20:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nlgtm","accounts_in_message":[],"_revision_number":3},{"id":"91237a8cb9d1438ac0e0ddf1dd74197ff87b9ed8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 13:21:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a3f6ad7cd5f8110b50b96633bf666643fc03e56a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 17:20:10.000000000","message":"Patch Set 3:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"41c634647e36f93ae267b0f5a47cebd1a20ecb06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 02:39:21.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78745c3bfaf54ed0a69459c645fe213b : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/293f99a7d02e4f64964782e214884ae4 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a541dd77d75460b9dabcd30a9686cd5 : SUCCESS in 8m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48033654c3de4697bd21f4da9b023f01 : SUCCESS in 6m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac3d6173e1db4b73bc36c78aaa803d3f : SUCCESS in 6m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1a9b3c7749ef41ad827ce42d19ab9be9 : SUCCESS in 6m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1047323c1a82459dae8e6a1008655113 : SUCCESS in 46m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34da3082b02a4ef8b865e3bc97ad757b : SUCCESS in 1h 28m 43s","accounts_in_message":[],"_revision_number":3},{"id":"3ca1660b0cc1c2fc05c1c8703c6ebb577d6bab97","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 02:39:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c55fe34924cda676a3f42eccf46ef1624c5d248b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 02:41:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e4e91dcad4b41f6af0dd9cc9f645a92 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ecf18825abca7660eb2478c7daa3f785bb5513b1","revisions":{"e32e39ab6f73b5ba09f11b8d775075225c3d3772":{"kind":"REWORK","_number":1,"created":"2021-06-23 13:54:39.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/54/797654/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/54/797654/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/797654/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/797654/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/797654/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/54/797654/1"}}},"commit":{"parents":[{"commit":"269a7bc5ffba8a0830c20fca6a31a45612c9c3f6","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/269a7bc5ffba8a0830c20fca6a31a45612c9c3f6"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2021-06-23 13:52:16.000000000","tz":60},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2021-06-23 13:54:17.000000000","tz":60},"subject":"WIP: Make tox -e generate easier to use","message":"WIP: Make tox -e generate easier to use\n\n- auto detect python paths (which)\n- make it possible to call `tox -e generate` w/o extra args\n- update list of pythons to include currently supported versions\n\nWarning: this will only work when run under Debian/Ubuntu, as on other\nplatforms like RedHat installing python-nss from pypi does not work.\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1974890\n\nChange-Id: Ib727242c1aba8597389259d8dbacba7194f9f0d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e32e39ab6f73b5ba09f11b8d775075225c3d3772"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e32e39ab6f73b5ba09f11b8d775075225c3d3772"}]},"branch":"refs/heads/master"},"a7e5a35a628b7047f739797eb8467809e3827f2b":{"kind":"REWORK","_number":2,"created":"2021-06-23 16:34:25.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/54/797654/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/54/797654/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/797654/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/797654/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/797654/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/54/797654/2"}}},"commit":{"parents":[{"commit":"269a7bc5ffba8a0830c20fca6a31a45612c9c3f6","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/269a7bc5ffba8a0830c20fca6a31a45612c9c3f6"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2021-06-23 13:52:16.000000000","tz":60},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2021-06-23 16:34:14.000000000","tz":60},"subject":"Make tox -e generate easier to use","message":"Make tox -e generate easier to use\n\n- auto detect python paths (which)\n- make it possible to call `tox -e generate` w/o extra args\n- update list of pythons to include currently supported versions\n\nWarning: this will only work when run under Debian/Ubuntu, as on other\nplatforms like RedHat installing python-nss from pypi does not work.\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1974890\n\nChange-Id: Ib727242c1aba8597389259d8dbacba7194f9f0d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a7e5a35a628b7047f739797eb8467809e3827f2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a7e5a35a628b7047f739797eb8467809e3827f2b"}]},"branch":"refs/heads/master"},"ecf18825abca7660eb2478c7daa3f785bb5513b1":{"kind":"REWORK","_number":3,"created":"2021-06-24 09:29:53.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/54/797654/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/54/797654/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/797654/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/797654/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/797654/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/54/797654/3"}}},"commit":{"parents":[{"commit":"269a7bc5ffba8a0830c20fca6a31a45612c9c3f6","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/269a7bc5ffba8a0830c20fca6a31a45612c9c3f6"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2021-06-23 13:52:16.000000000","tz":60},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2021-06-24 09:29:44.000000000","tz":60},"subject":"Make tox -e generate easier to use","message":"Make tox -e generate easier to use\n\n- auto detect python paths (which)\n- make it possible to call `tox -e generate` w/o extra args\n- update list of pythons to include currently supported versions\n\nChange-Id: Ib727242c1aba8597389259d8dbacba7194f9f0d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ecf18825abca7660eb2478c7daa3f785bb5513b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ecf18825abca7660eb2478c7daa3f785bb5513b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
