)]}'
{"id":"openstack%2Frequirements~798722","triplet_id":"openstack%2Frequirements~stable%2Frocky~Ia3346187e13acb880b3447598012f1530856f5c8","project":"openstack/requirements","branch":"stable/rocky","topic":"cap-hacking","hashtags":[],"change_id":"Ia3346187e13acb880b3447598012f1530856f5c8","subject":"[stable-only] Cap hacking","status":"MERGED","created":"2021-06-29 17:15:28.000000000","updated":"2021-07-04 03:22:30.000000000","submitted":"2021-07-04 03:22:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"798722-cap-hacking","meta_rev_id":"43503fd92db2ffaaf926a180c526c157da1585ba","_number":798722,"virtual_id_number":798722,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-04 03:22:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-07-04 02:11:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-07-04 02:11:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-29 18:45:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-03 22:33:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-07-04 02:11:00.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"30b230374dd2f1cf40cf7a95df32d995eb595597","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-29 17:15:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10bdcc4ad9c9e760fd852196feeee66b74c29c2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 18:45:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19edcaa1b5fa4c82a180d63d1c878f6b : SUCCESS in 6m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0926a116d53941adb74fbb68e275ec35 : SUCCESS in 6m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/60ba0a0c25cb47fcbf599682e06e2761 : SUCCESS in 8m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8117a1bcf49d4e9cb3566d080f164a41 : SUCCESS in 8m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b70f92c525fc4186a8e879b5bd524b64 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f171ab75d3b6408a8c3bb41df18d567b : SUCCESS in 6m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0a63beb4fbf24c4e8f54bae1a07389ce : SUCCESS in 58m 10s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/7fabc50f15ff42869cc303397f38ec7f : FAILURE in 1h 16m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1b7994582f334e05b1c011460a89b130 : SUCCESS in 53m 07s","accounts_in_message":[],"_revision_number":1},{"id":"9bf6e1d0a6edb64f06757d0c01b5fe4be427c797","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-03 22:33:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8f13fde0f6a34931ba6d8e70bd3fd325610f32fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-04 00:44:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71f60e645a8b4421a85cf3bf2a745e5f : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/be0f49f2d98342a687f7345577edf0cf : SUCCESS in 9m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/826f55c874d042c0a2d2cee5a81aa80a : SUCCESS in 5m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bc97741004ca495fb4ae70870a888fbf : SUCCESS in 8m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/046499e7fe1d41f8b53e09a4ebf1e065 : SUCCESS in 7m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/577ff996339f40a8a0fd62e9f6227580 : SUCCESS in 12m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/df04b2ff7e6941ddbae5ae0dbacd6637 : SUCCESS in 48m 47s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/ab03f96a0ca74aa3b96669200b86caa9 : SUCCESS in 1h 35m 09s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e5614ff92d1c4007add05c8f25f1bc1f : SUCCESS in 43m 54s","accounts_in_message":[],"_revision_number":1},{"id":"e60ee8e540816198ecc31bb8cffbd9b4a85b5f4a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-04 02:11:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8b926afa04e9617d7f940c8761c33ef27760072c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-04 02:11:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1322dbf0aa1d55aa5c77abe87cc76a999abec01c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-04 03:22:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db953d9832664deab8533e574c1f6531 : SUCCESS in 5m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/98a54767eb83452f80984d30f0b8267d : SUCCESS in 4m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/860205cf06b341c3bb1a6b3761fcb6bd : SUCCESS in 5m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/462c3e21a5104821bc0e905cf2b172ca : SUCCESS in 5m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8299835d3d604be2abbb8cc6dec0a81f : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/137ab8609c5442b28bd8e006510906eb : SUCCESS in 4m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0453753425f942afbd5f98fa2a00e1ff : SUCCESS in 50m 29s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/d1e53689db3643a48127493d8bee9547 : SUCCESS in 1h 09m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d01fc5ed685841d89c4686b402534a28 : SUCCESS in 44m 59s","accounts_in_message":[],"_revision_number":1},{"id":"43503fd92db2ffaaf926a180c526c157da1585ba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-04 03:22:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"adb8d5da24a4c5e9b1087f6588616c43961bbcc6","revisions":{"adb8d5da24a4c5e9b1087f6588616c43961bbcc6":{"kind":"REWORK","_number":1,"created":"2021-06-29 17:15:28.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/22/798722/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/798722/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/798722/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/798722/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/798722/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/798722/1"}}},"commit":{"parents":[{"commit":"c4e0d76a0c9888bbcfa973ca15673b8120971ae2","subject":"Ensure virtualenv is installed locally","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c4e0d76a0c9888bbcfa973ca15673b8120971ae2"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-06-29 17:00:39.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-06-29 17:14:56.000000000","tz":120},"subject":"[stable-only] Cap hacking","message":"[stable-only] Cap hacking\n\nWith the latest hacking release, 4.1.0, the pep8 test started to fail\nwith the following errors:\n\nH216: The unittest.mock module should be used rather than the third\nparty mock\n\nThis patch cap hacking\u003c4.1.0 to unblock the gate.\n\nChange-Id: Ia3346187e13acb880b3447598012f1530856f5c8\n(cherry picked from commit b917ca42100572e4af6896d20a47eb806e16acf1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/adb8d5da24a4c5e9b1087f6588616c43961bbcc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/adb8d5da24a4c5e9b1087f6588616c43961bbcc6"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
