)]}'
{"id":"openstack%2Frequirements~799048","triplet_id":"openstack%2Frequirements~master~Ib6918115c45769ef81a21d6212bcfe03b0fb0ad6","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib6918115c45769ef81a21d6212bcfe03b0fb0ad6","subject":"update constraint for oslo.log to new release 4.6.0","status":"MERGED","created":"2021-07-01 12:41:49.000000000","updated":"2021-07-03 08:25:49.000000000","submitted":"2021-07-03 08:24:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799048-new-release","meta_rev_id":"957209da715388893732adc2f4008279eafbc876","_number":799048,"virtual_id_number":799048,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-03 08:24:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-07-03 06:58:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-07-01 13:09:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-07-03 06:58:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-01 13:06:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-01 13:09:45.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-07-01 14:13:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-01 18:30:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-07-03 06:58:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ce3d01765f59675fff45a70166d489725518f329","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-01 12:41:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4da801169d9c017b2bc9d90a6defa3632f84914c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 13:06:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fd0653338ff34d05ad9d7ddef949891c : FAILURE in 21m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d4f4563a8423e866ea8a4a50c15d27299eaf51a8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-07-01 13:09:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b06f769c5ef656acc2c849a96e0b45acc57926ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 14:13:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/807ce509405b4a41aa42820f3c939ebd : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f41f14bb7d7345ecaf4d95cb1f293ada : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7220a23116ce4b9f99fd4e2c1d450a6c : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f693dc8eb664637925fbcd136bb3034 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f04fc33c64f4de591154e8384a578e8 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/98d5555804014528bfa64470e9d6258a : SUCCESS in 5m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3760ff9dbf5e473b9d550f81c6cc209e : SUCCESS in 5m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/39550dcd134547b284f8ba12983659fb : SUCCESS in 15m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d4a3116165ab4c9eaeea810e8d132a94 : SUCCESS in 11m 12s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ba0bcb2f40b24c828fb2696c6e9d0303 : SUCCESS in 6m 44s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e893377554f442869d695c1df75eb6fb : SUCCESS in 10m 35s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/af8d8a4ee8e943ea89dad3d6ac4c82dd : SUCCESS in 6m 41s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ddec88c6ae3c4c55957c092cb59f7fa1 : SUCCESS in 9m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/103a5d8a83614f86a0a4935ad88cc1de : SUCCESS in 8m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7ce614bc7c714d4e84b5031910071463 : SUCCESS in 10m 01s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1aeaab0ff7684234bac9216221124dcb : SUCCESS in 19m 55s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1e591d17b46a4349885e1689fb86734d : FAILURE in 11m 57s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/694e1b64669d4dc1a9897bd5d6907375 : SUCCESS in 6m 39s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/aeb739ede10243359e95d8eb480a5b09 : SUCCESS in 22m 01s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d855479f13354028839e1034c2f91913 : SUCCESS in 13m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d460601e93744dbeb8c31e9991ab2cdc : SUCCESS in 4m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7c577edd9a254f4388e6c58966646de3 : SUCCESS in 20m 43s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/521d565ee99c4c539c76070006463a5b : SUCCESS in 6m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b7c8cc2588a54b8a88f3bb726099f6fa : SUCCESS in 7m 45s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7853d0838a0f4d93a5a5988a00f88ab2 : SUCCESS in 14m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/564ca9a2b9794b59bc66a96ef79cb22f : SUCCESS in 46m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0c698e74f84243e896978fcf4cb7338d : SUCCESS in 1h 30m 14s","accounts_in_message":[],"_revision_number":1},{"id":"ebb77326e355a54e21c2cde8f4ab88128a38ba38","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-01 18:30:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"945c7cbc84a229df132707676310f2c785bea418","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 18:52:06.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/109bd96c9fe64ffcad5013a7c283893d : FAILURE in 19m 23s","accounts_in_message":[],"_revision_number":1},{"id":"53e0b6e4b12b8fff4f5fdd2c2f89ed39fd12d944","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 20:17:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c31bc84006694d469caf4296efca1b62 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7460a9ac8784e8abc13e3ae84ca4db7 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b49333560fc433d82b499c68d7fc170 : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21ee3061e9f24e66b89a3c8b9c491e71 : SUCCESS in 7m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da0b0ec6d24b43be86658c76b36ec54d : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d8451269cfed40d08de3a489636224ed : SUCCESS in 6m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/06918889b4374fc6ba0f372907cccbc6 : SUCCESS in 5m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/66156daa615b4cd6961e24f21dbf4ef0 : SUCCESS in 12m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5fd21fd5e58b4ba3affc826e49efc360 : SUCCESS in 12m 11s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/185874f22adc4dcfa702762b6490d1a3 : SUCCESS in 8m 18s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c50cc9f7d528498ea350255e22eceb3a : SUCCESS in 13m 06s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/37ccea0a16f04cc6b9ca4c02e6757154 : SUCCESS in 9m 46s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e2d14ce29af14d5abe89167142242f72 : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7231bc9b52c543ca807c6988593d2624 : SUCCESS in 10m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/31d0f13566ad434ea6cc376811150818 : SUCCESS in 12m 25s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c98a0ea6394844e2a5205f6a5601c628 : SUCCESS in 16m 31s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1c0034e3e3af4394ba99bee0bbf5048a : FAILURE in 12m 35s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/346d2cc3079f436ba06ef410accc8e45 : SUCCESS in 7m 39s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6e0e43acc8d54bd790703d2b44f79fc6 : SUCCESS in 21m 29s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/520280ffa73c4dcabd3738a4413577dc : SUCCESS in 14m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/49a2beb65b054fd0b0b61cc790605f2e : SUCCESS in 5m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/154c7f3ca9a74fe6b2f5956128c49231 : SUCCESS in 22m 36s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f99062064e2a497d8b8af76d7603bd61 : SUCCESS in 8m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ba21ac62e4e94c46b1aabeb7bb64f559 : SUCCESS in 10m 27s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/65384e70b9124a13b9b3ea7610a7af65 : SUCCESS in 13m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e34d24879e0548b2aca67fe77b252c1b : SUCCESS in 48m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c38ae62cd550427eafe12b892bc47d49 : SUCCESS in 1h 40m 23s","accounts_in_message":[],"_revision_number":1},{"id":"3a980b433269ab88c2456a0ed287f9ab2160b1f9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-02 01:21:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0e50b7576a014e782186bf60a3d2ea49952423f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 01:39:43.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/aa78f6f2c48e48409032a5a9b7a32d06 : FAILURE in 17m 00s","accounts_in_message":[],"_revision_number":1},{"id":"a89b40f57774b4cb10c18b9296766e72377b5386","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 02:57:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfe86179b48e46af8f0cadf3dce7d393 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d76a224036a74c8b9f36e1a8ca2fd8de : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a5dccce4c214cf99711f71e8e7cfe27 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c79da4d599ad473ea4845c8fb9ec88a3 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10a84371c40a4956962a5844fda2c6e7 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/baf13d90c8be446c9d91d1e633343b5c : SUCCESS in 5m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6d7aff5e40ad451d9770de0ee8e4209b : SUCCESS in 4m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2fd0dd7f59a244148988d34f9e743209 : SUCCESS in 20m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5f554508ed4b40bf91082efcb0fd1c91 : SUCCESS in 12m 18s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/9b3125a933d0421a8f49067e97fac4a7 : SUCCESS in 6m 45s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/eb6db778222f4113914db774efe5f845 : SUCCESS in 11m 33s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/1a00cd3ebc5645ef8e09a8323e38af48 : SUCCESS in 6m 01s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c701de27bebc4b678494a29e1493366c : SUCCESS in 11m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bc0e1de6eb05441fb21027e3000aa4eb : SUCCESS in 9m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b7616b2188db443bbf76b7eba5ae4fe0 : SUCCESS in 10m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/51dcfdb1d3ff4c3389b9955013796833 : SUCCESS in 18m 17s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/3de08f29bdeb491aad0fe1278458a679 : SUCCESS in 9m 09s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/470620eaf04d4430992e91d8045b6d64 : SUCCESS in 5m 42s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a1de095ddc234aac869d5bea7f0e46f6 : SUCCESS in 25m 58s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7b13299ad06d4c37b4d99fbc93e82472 : SUCCESS in 13m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3d4bb543d656447093c13f9df9e966da : SUCCESS in 4m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8c6ade5334f14fd5b7531595cd118f55 : FAILURE in 25m 10s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/12e5ff9c040444f299e0ef1e65ff47a5 : SUCCESS in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d445d1ec2d7149dc94280790389139cc : SUCCESS in 7m 29s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9cb50d4485da4d7f8d81472de1fee05b : SUCCESS in 13m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0134764d46f64554bd0ab4ba96c2c59f : SUCCESS in 53m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/615ec6338f904b2ca7fd2094073aeda6 : SUCCESS in 1h 30m 26s","accounts_in_message":[],"_revision_number":1},{"id":"955b33a63d89ca2e3086fb21493f9f32ed38f894","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-02 15:52:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9676e5ffba7ff494286bc166e81bc1d2e6048413","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 16:59:19.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3e0fced8d4a64e4584a351e60a5e85d1 : FAILURE in 18m 55s","accounts_in_message":[],"_revision_number":1},{"id":"7ec8c1afcfc59942fc0632ab9a40867f71ad462f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 18:14:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d137fadd22d64b45adb848eb131404bd : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eb596e35b9d41ccb69df98ac4ad63d0 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32aded36db0c4ce28f0104af3113be44 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7de700dec9441f8b2c85683af4ef695 : SUCCESS in 7m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46d5bc91c8c943d1bacc0db7dfe0a96c : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/82e63e1210b14206ad59cbf24aa68947 : SUCCESS in 6m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b4b76742ed1f4ff89cfe197b3643a668 : SUCCESS in 6m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f23c05e245204daab0d19a801fcc26f1 : SUCCESS in 11m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/57ccd451665d41cc984c5b105ca4eea1 : SUCCESS in 16m 48s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/3ed4db08341b4287aef2b664b3edbfbe : SUCCESS in 5m 50s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ab5d16271d9c432cb8bacfff031dc25f : SUCCESS in 13m 14s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/edd17ae418294084b34ff2ce000a81ee : SUCCESS in 9m 13s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7aa484542e4d44a78ead530547173470 : SUCCESS in 12m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2f7d5f3d11a746508de6fa52509091c9 : SUCCESS in 9m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f544fa15fa7d470696eb4f45c2498c06 : SUCCESS in 9m 58s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2666be31241344619d98f59b88c90b8c : SUCCESS in 14m 21s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a2f85820a2db4463a2404b836b9bd0bf : SUCCESS in 13m 13s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9d6f46e17c8b43a6b5eb503108c16e60 : SUCCESS in 6m 43s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/285a21af99594f999ff3d9d4fc565c58 : SUCCESS in 27m 39s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a775b41e47dd4250b7f444101681b11c : SUCCESS in 15m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2a4f346498f54c1cb22124787f04bd54 : SUCCESS in 4m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/079b7f384725488fa6f8384c0bbfc6cf : FAILURE in 22m 03s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/11890335fa1f452bb657833e5b91f015 : SUCCESS in 5m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fc098baa22e24e8e917ada232dc090ae : SUCCESS in 7m 45s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/737bef0256cc4bf2ad85f4b8fd41b988 : SUCCESS in 13m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59c22917819547d1bd27fc3282c96eed : SUCCESS in 50m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1429973a45d141079372c89d683caf02 : SUCCESS in 1h 16m 36s","accounts_in_message":[],"_revision_number":1},{"id":"a6a29d9e366307731f29e313b760d2b2a61e6176","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-03 05:04:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1ea77101a6c8693522ceae51ca20c316f6fc6b9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 05:29:43.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2c40c9c2aa694a3bb806c54a946db20b : FAILURE in 21m 16s","accounts_in_message":[],"_revision_number":1},{"id":"284dc6b625266a904072d21a618eb63b77e10a09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 06:49:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8644cd8c11d4425b94733a62f033109c : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c66f61dd3dc641d0879cf4b6e5146712 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc0ef09508824089b8cf7f1f789731ae : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3a463dffb5e4f26a0ec0b079821444a : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be8141e0bf8c4a0c95789fb5a31ae6c2 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e4f0948b03bf4a929bce3eccff0ec479 : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9de6b8b795484b839e8f175c817f6a51 : SUCCESS in 4m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bd43ba3a5d194a829a0a742a6f6d53c9 : SUCCESS in 11m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8558bce450f043ec9d762ee330025a62 : SUCCESS in 11m 22s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/890ad96048d54e9b9d048fcdef226ef8 : SUCCESS in 5m 43s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7650274e33834dacaf28086a2c3fc53b : SUCCESS in 10m 26s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d48f4dd1b89f430cbe37934bcd80abae : SUCCESS in 6m 11s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d3fd91e7bbcf435c9e5150d328c5b769 : SUCCESS in 10m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/73484988c30648c6a2d6c4099f92d39a : SUCCESS in 9m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/65baa84472aa4b90a643f4680cd8c679 : SUCCESS in 10m 04s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e9e9a1840ac94522a318f4643f91408c : SUCCESS in 15m 30s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e62cee61a84e46d9bb6d3e0908168ca0 : SUCCESS in 9m 22s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ab62428fdaf945c89c3fe3b60f597df0 : SUCCESS in 5m 21s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3eaccb5451c2432b95e3b37aab83ea13 : SUCCESS in 20m 57s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b1cfaa65d7514c8e86dc502640ffd1d3 : SUCCESS in 12m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3f1cfa8182c04a628a88d7940b4880fd : SUCCESS in 4m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/46403f9c8f04444da005a49709ec9e48 : SUCCESS in 18m 44s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/97097121198e4eac8b10dea787eb6c64 : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0608d8c362084de3bc2e61bb69c3c0b4 : SUCCESS in 8m 32s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b5b52f66b9df4fe5ba43f7da781f36ca : SUCCESS in 13m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69fb387ccd9c49eb856a8dcf0dd53147 : SUCCESS in 45m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4f942b073edb4e88968aa64ea999384e : SUCCESS in 1h 40m 15s","accounts_in_message":[],"_revision_number":1},{"id":"e728bd4c140f575700f0ca8efdfc7dc1971d0b78","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-03 06:58:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9d1bb3d8e82ade786d07475cb01f9101e109d677","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 06:58:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85a9ae2e0c96f6897d2bfc313884781dbca347ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 08:24:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d31202de7484499583dbc528d7f2ba25 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a52976eac3834491b5c50f3699f5cf74 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06884e5f51174fd880c681b4e4fffedb : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77148166bee74159851f1855dfdf4c12 : SUCCESS in 4m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/31575d51669b482dbae5e7715ee4090f : SUCCESS in 4m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bf5cc79d3c644c2496a208bb232b2bb3 : SUCCESS in 3m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/aa4c57bed3ad4a1a8c3996fd01ce28eb : SUCCESS in 10m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cbb4d5ab2b6c4c5f97d70b09b4d04e82 : SUCCESS in 11m 38s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/afd3659e21a741cc87eaa77cafca5916 : SUCCESS in 18m 50s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e728a85749e94f108f0fb7f6ee79df2e : SUCCESS in 6m 28s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/57568d95fa3f4bc2a969e97abcf82f0a : SUCCESS in 11m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7dd81fa2858844b0a387dc853b2f5257 : SUCCESS in 9m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/23a953cc4fe345eb90b53f988af3e196 : SUCCESS in 9m 51s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d265b8dd4aa94933ac5332d7395c7729 : SUCCESS in 13m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8a9d2cc4490c463cb008734eecccce5c : SUCCESS in 9m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/43fd3e3cab584029a9539aa827e7ae20 : SUCCESS in 5m 53s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/16d94d8dfb664d69adbafbf01492e206 : SUCCESS in 25m 06s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/283ea827dfb9425d8f99a1a64462c723 : SUCCESS in 15m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f1d8cc3158d340d0a453d036612714dd : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9d159b56506d41ceaab514c095d77d13 : SUCCESS in 19m 50s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/bb298ac9f19047fcb6209ef921221bbe : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/abec24e5760746f2afcea4962de19980 : SUCCESS in 6m 59s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ff414a93ad354ac0981ded0fe5292eee : SUCCESS in 12m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4419555b4a774595baa7159bf7110e89 : SUCCESS in 47m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da62ac3b3ac74ed58b4f50585de2ede1 : SUCCESS in 1h 21m 50s","accounts_in_message":[],"_revision_number":1},{"id":"1b1345ac56d0718c6a4adb0e8aba9f7aa5fd2680","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 08:24:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"957209da715388893732adc2f4008279eafbc876","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 08:25:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7764342d669441b8fa735e79c8194d2 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9bbfdccd2a9bc30c305293c2e961a0dfbc8feaff","revisions":{"9bbfdccd2a9bc30c305293c2e961a0dfbc8feaff":{"kind":"REWORK","_number":1,"created":"2021-07-01 12:41:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/799048/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/799048/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/799048/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/799048/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/799048/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/799048/1"}}},"commit":{"parents":[{"commit":"5c4901e38e5ee51cd28cf6e42ca904f19a001f47","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c4901e38e5ee51cd28cf6e42ca904f19a001f47"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-01 12:41:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-01 12:41:46.000000000","tz":0},"subject":"update constraint for oslo.log to new release 4.6.0","message":"update constraint for oslo.log to new release 4.6.0\n\nmeta: version: 4.6.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ie08cda2ed6285c9ce506432cb516161a85b49076\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ib6918115c45769ef81a21d6212bcfe03b0fb0ad6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9bbfdccd2a9bc30c305293c2e961a0dfbc8feaff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9bbfdccd2a9bc30c305293c2e961a0dfbc8feaff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
