)]}'
{"id":"openstack%2Frequirements~799050","triplet_id":"openstack%2Frequirements~master~I77aebc0ce8bca93d3c04c9ddf10e6c6d98d5fa9e","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I77aebc0ce8bca93d3c04c9ddf10e6c6d98d5fa9e","subject":"update constraint for oslo.reports to new release 2.3.0","status":"MERGED","created":"2021-07-01 12:45:17.000000000","updated":"2021-07-03 19:10:10.000000000","submitted":"2021-07-03 19:08:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799050-new-release","meta_rev_id":"492c35a83d127681db6cf2ab8a73b8378c31c557","_number":799050,"virtual_id_number":799050,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-03 19:08:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-07-03 17:48:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-07-01 13:10:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-07-03 17:48:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-01 12:58:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-01 13:10:03.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-07-01 13:57:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-01 18:30:55.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-07-03 17:48:02.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"9da5010c7b9047547fadc94187f383d38a3628c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-01 12:45:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f51bf7ea961387de330e2e15ceae03c68ae696c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 12:58:14.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ac2cb744d13845e6b5d65988eed6c545 : FAILURE in 10m 43s","accounts_in_message":[],"_revision_number":1},{"id":"432e0b713319fcdfd67e1041ffa0f9cbe8a8f9b9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-07-01 13:10:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ff160c38c64a98be07589293233342f07d6bfd94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 13:57:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b894150451ea4817b32c43ca4b1033f7 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/018442ea903f4a4ca3493533fdcaf095 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61e0415ccaad4db5b5c5912e5e69b0d6 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47d4a0150ee24d829bcc0272c1bdcb29 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87b86fc2ed334f42858eb68af145a672 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/02b45ed92cba4329b9c165323cb94d11 : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9c100bc591164a489b7aac2a82932ea8 : SUCCESS in 4m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5fae121bdd4e4fd3a605ad19212d2bb0 : SUCCESS in 12m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cf8e1066f0084bc08989fe140d5666c4 : SUCCESS in 12m 46s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ffa85f4197854da499e1bcc432dc2c37 : SUCCESS in 5m 48s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/77199bc7fdbb4e9da1f091c8bbdc9345 : SUCCESS in 10m 41s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/bedb98bb0aff450cb354923e85d082df : SUCCESS in 6m 44s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/236a6571d4ef43b498d8ffd0a8b0368f : SUCCESS in 11m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8f689db65cdb49bbab7aa6086ccd647b : SUCCESS in 8m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/96f06d2f7a7f43f5a709cdafc8bffda0 : SUCCESS in 9m 55s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/252c09eebb264c75a1ed8d55913ac0ec : SUCCESS in 14m 12s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ec89334a5101467886f56952a786e842 : FAILURE in 9m 32s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/185c69fb2e2040dd92c8c4ca9a16d5ec : SUCCESS in 6m 10s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8b33691448964ebab0fa67af2662bc55 : SUCCESS in 28m 02s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8193f2f5796d477eb8a24a515b360972 : SUCCESS in 13m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3c14f6cabe1b4a398ff2df445ca91515 : SUCCESS in 5m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2b698598f40a4cb5bb0bc67701e4019c : SUCCESS in 21m 36s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7d319466704541bf86477fda9b0a8b60 : SUCCESS in 6m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ee00c4d265ee422a9f8e814a81565872 : SUCCESS in 7m 09s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0789ddd32adf48f7a64000f3db4c7f2a : SUCCESS in 12m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38340fbf4f2f4d13bfe0425a66183b88 : SUCCESS in 45m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a537eb3e12dd4602a01c23713ca1b540 : SUCCESS in 1h 09m 56s","accounts_in_message":[],"_revision_number":1},{"id":"55c3b31b1332f995e8875ee04531d43a143441c7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-01 18:30:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ea57b6ba0740899f76666346e6d54678d38e1aaf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 18:51:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/853c5a74d29441aa91ef3c38d045a2b6 : FAILURE in 18m 53s","accounts_in_message":[],"_revision_number":1},{"id":"93d90ae6d4324b57a8f7cc93615d6bb59f268cb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 20:51:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fca1af1b75a4039b364c81bff7f51fd : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1f40ef034b84b6995a73bac73d85e2b : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/621c585026d24407837496fb514c3985 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fe4c6b7ad874484855795045f06811a : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18ac289ae9ec4a459a69fc2aa78e91d5 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2625e9f0c8ce46c0b7f682f669ed9e51 : SUCCESS in 4m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e8715cff8d214c3c84a8e6bb52b314b4 : SUCCESS in 6m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c6ac25c4a43f45be908bf18a67d818e6 : SUCCESS in 12m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2bd9bdec93fd474b85ba3514c248b392 : SUCCESS in 10m 50s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b342beadd4234f75a1b02c3578bc6c81 : SUCCESS in 7m 48s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/bbce0082623a4d64a67f0b3ad41e9810 : SUCCESS in 11m 22s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/bf79933d51a24e56b8904b665e4e18fd : SUCCESS in 9m 01s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1dd784049eda4556befa4b80f3c08618 : SUCCESS in 10m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/50b9f0b962324bbeb7e2062bbffe8bca : SUCCESS in 10m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7de8545de13e469fad512a6d9957b885 : SUCCESS in 11m 25s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1ae82bd58e0d41fc8cc7ec7974f22732 : SUCCESS in 19m 12s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/48189c2b76d842e5b34cd90347cd1d30 : FAILURE in 12m 45s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/09bd309b647a4a4c9caed95920a25221 : SUCCESS in 7m 51s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6580906361d744c9a028eef7c6c29e64 : SUCCESS in 22m 50s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/226403cfda204063b3e8484f813f8058 : SUCCESS in 16m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cad437c6deb04d8aa1a02bc85bc6cac5 : SUCCESS in 6m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b3cf5c6127ee465cbded4a0266183181 : SUCCESS in 21m 23s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e4be1a85e5834570bb4bc4326228faed : SUCCESS in 6m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cdf1d44eb9a7488084986e1b9d57573a : SUCCESS in 9m 05s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4224da17cfba437e844b43c93367f229 : SUCCESS in 15m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0bac02536514fba93cbd2064e422e65 : SUCCESS in 53m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/85314595228842ba8a3c59a05f49f317 : TIMED_OUT in 2h 09m 49s","accounts_in_message":[],"_revision_number":1},{"id":"53b29fd0c7ffaeb5584d753ebeea9222b39bc058","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-02 01:21:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3ba7fc3afeac1d30595a63727d5f94e59e99b89f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 01:39:44.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0a8d90d6609a4cc89e2f08de424f10e0 : FAILURE in 16m 50s","accounts_in_message":[],"_revision_number":1},{"id":"8dc269539bbb3920479ef0a4020af0923220711a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 02:44:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a86973d7e55943e8a2ea99796070feda : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/523670bc377e4c9f95f8954596759c98 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1a50f5cf4da4150b097ec26049895e9 : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1714631baaa44eff84d95143690bb72a : SUCCESS in 7m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5da9bc490db84d49a217806e59a4a2da : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/009fe0597561424d80356aeabc6abaca : SUCCESS in 6m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/51dad46f6c7a490da21caabb06492cf7 : SUCCESS in 4m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/030755fedfa04742a5d41537795e40a1 : SUCCESS in 14m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5d00433b57844ecdbad3dc3c21f268cd : SUCCESS in 10m 56s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/faf189361bd843608f0d8a1e3292374a : SUCCESS in 7m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/84f9fa0c53ab4a80964e2a511ad1e69e : SUCCESS in 10m 23s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2e065d3018604fb59676e0a3a8f2b10c : SUCCESS in 8m 32s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/044c0bd109354d6b8c2e5d878f2cfe94 : SUCCESS in 9m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f74abf98ff7b4864a08ddaa9e378ca6d : SUCCESS in 10m 02s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4d3b35051ec846a3a7720e92d1688bf0 : SUCCESS in 12m 40s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8148038b759d476daa7fec5f1ca77bc4 : SUCCESS in 19m 05s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a508e9652f0b46078ea97d56d399af67 : SUCCESS in 9m 37s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e816ecd6d0334ed18f14349d7cd1fa25 : SUCCESS in 7m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/dc8311ff7dfd4a10987b1ed83076cb0a : SUCCESS in 24m 34s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/725209bb04fe4f56a288608fac6db619 : SUCCESS in 16m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/566fda54f2684a37bbc121051ff0102e : SUCCESS in 4m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8483dda2508443f7a3aa7f8ad41166f3 : FAILURE in 24m 25s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/24914a2763ae4acf8df57dc07eb7fda5 : SUCCESS in 6m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/798709ccc01141e69e9f78708d2ed7a3 : SUCCESS in 7m 16s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/448395bb0b34492b9c3bba0d7442bf3c : SUCCESS in 12m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05f7dd263b704be2bbb477b8e46c18b2 : SUCCESS in 52m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05a2c2a1db454fbc817aa3f6cc6f5c16 : SUCCESS in 1h 22m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a983703c90c04f9e6b87fe8c4c8aff8db11ca34c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-03 06:58:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8cb94c2270aca3644903f6de8e3179c40492ef04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 07:29:56.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/16cd96593b2643ca87efb3add30c9b62 : FAILURE in 19m 57s","accounts_in_message":[],"_revision_number":1},{"id":"35446fff83f1488f38cc8de0c70fa3633881538d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 08:16:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3ab620518fd4cfdb51a3fc2ca962cdd : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65dcb7bc51f94115b801a200a60bf774 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddf820813d2d4e7382dc9ff2387e11bf : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23a8c9b3a3d34aaf9626307fe3e6fb6f : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f658df6497804e1c9a9b4782219c4034 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6be02ae9c02b4eca99fbaf3aae998185 : SUCCESS in 5m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2550a01d32014d78b352e7a9185eef9e : SUCCESS in 5m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2d7754a6f46146289fdaaebec5254cd7 : SUCCESS in 13m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4c76ece3c9e54751967c9a4ca080a2a4 : SUCCESS in 9m 51s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/4c8017d04c0649378d86294c5d9b8bcf : SUCCESS in 6m 53s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/1ecf7f4d318c40d3abfc906e487b042f : SUCCESS in 10m 38s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2640c652fb134d59bb10ba4bdc1fea98 : SUCCESS in 7m 18s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/10ad644a97f54d7f82db48be83236118 : SUCCESS in 9m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e6b9f56526fe4ea3ba538bcf29f0acb9 : SUCCESS in 20m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4ca89454eb16485cbdc62d61446e34fc : SUCCESS in 9m 38s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0bf2047a271a488397f9692d1528132b : SUCCESS in 14m 03s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4facb07314a2420997f6cd6ddef318af : SUCCESS in 10m 48s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c19fa87fbbc7469fb1fe6460ed881fdf : SUCCESS in 6m 03s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8c7da4bfbc7743ce8ffdd277d1ea9b1d : SUCCESS in 25m 35s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f89df5e1e2004675910dbf9d941b833e : SUCCESS in 11m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3dc95d4f9fe448198d7442cacb78b1c5 : SUCCESS in 5m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9f4875acb7814e9f9166226c182a34e8 : SUCCESS in 18m 32s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/cb5386e9fb294e93b87f4bebcb113515 : SUCCESS in 6m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ab18862b200142e0b11f4347e0039cb1 : SUCCESS in 8m 14s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b9f9dad9431d405c8caa1745532f67ad : SUCCESS in 11m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/08646f79af0e461aa34689bc5caffd8c : SUCCESS in 44m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/898af52e7cf44f27bc20cf0169f5f740 : SUCCESS in 1h 14m 28s","accounts_in_message":[],"_revision_number":1},{"id":"3c12da0454ebf2be760684dc929e2562d26a32da","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-03 17:48:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b8232305594f20dfed4b53f696080d154602a40f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 17:48:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2148b9d7ee2dfa94ecc83bcc531db3d0620c82ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 19:08:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4c8610b13d54e8bb69e658b93ca099d : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d94beadbef2b4d90bc80bb62cc992bed : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/caa6a9cb45fb4b05b8660c5cb1054b71 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fb98fdafa89411ab00c09b56e5b77d7 : SUCCESS in 4m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a3f5fe879480445e8dc96ccad9d29530 : SUCCESS in 4m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/30345c755a4b4a418664b0b91efbf898 : SUCCESS in 5m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fc77639d5a224a0190029552bac3c5a7 : SUCCESS in 10m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/40dbe7a0332e4250b0950e5dc7dac941 : SUCCESS in 9m 04s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/aded23a5eb024959877287d45f728d4d : SUCCESS in 10m 34s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f37aece3f1c140babc3f447c6d4fe4ac : SUCCESS in 5m 50s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ccdfb832c57c4d0d9b9ae46e50669e3d : SUCCESS in 9m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a634fddc3a104812a0fbfcb11689d38d : SUCCESS in 7m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/54ccbe80b137460d88ef5bbd3682f0f4 : SUCCESS in 8m 04s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e6615589a1574cdea9add9908e95aed7 : SUCCESS in 16m 22s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8258a15e0bca44118fdf171a637c4c2d : SUCCESS in 8m 29s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/70dfeeb418344cd79813bfcf5d657e24 : SUCCESS in 5m 05s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2f9c347949c546bbacbf138d8c90a719 : SUCCESS in 22m 47s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f4fa483ca1774613b5bc11e7994c9eb0 : SUCCESS in 16m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/37a107d4ceda4681880d34371c34f54a : SUCCESS in 4m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ab5a0b2f04c34d0bacda8aa7f0f187de : SUCCESS in 19m 18s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/38f8c80ed7f64b8e92852c70cab49bae : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e01a8c455c154e13838c65aceaa0ca0c : SUCCESS in 6m 35s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2462fae9e4a941dcaf67aa1e47cc3da9 : SUCCESS in 10m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/baa733b60ed84ef5aa7ce950a66d07bd : SUCCESS in 54m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/74c4c9ae43e04c85858d343fe2bca4a5 : SUCCESS in 1h 18m 16s","accounts_in_message":[],"_revision_number":1},{"id":"ce8eecf9df80884f5628630af9361c16f4bb9c84","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 19:08:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"492c35a83d127681db6cf2ab8a73b8378c31c557","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 19:10:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccdd2e80e6c842f5984f763cd33e4ccc : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"810a03ec019a1df48aabc4805732086358391d67","revisions":{"810a03ec019a1df48aabc4805732086358391d67":{"kind":"REWORK","_number":1,"created":"2021-07-01 12:45:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/50/799050/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/799050/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/799050/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/799050/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/799050/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/799050/1"}}},"commit":{"parents":[{"commit":"5c4901e38e5ee51cd28cf6e42ca904f19a001f47","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c4901e38e5ee51cd28cf6e42ca904f19a001f47"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-01 12:45:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-01 12:45:15.000000000","tz":0},"subject":"update constraint for oslo.reports to new release 2.3.0","message":"update constraint for oslo.reports to new release 2.3.0\n\nmeta: version: 2.3.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ie08cda2ed6285c9ce506432cb516161a85b49076\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I77aebc0ce8bca93d3c04c9ddf10e6c6d98d5fa9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/810a03ec019a1df48aabc4805732086358391d67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/810a03ec019a1df48aabc4805732086358391d67"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
