)]}'
{"id":"openstack%2Frequirements~799051","triplet_id":"openstack%2Frequirements~master~Ic55c0d8912e3934c4dab5bdce3d714a327afb081","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic55c0d8912e3934c4dab5bdce3d714a327afb081","subject":"update constraint for oslo.rootwrap to new release 6.3.0","status":"MERGED","created":"2021-07-01 12:45:50.000000000","updated":"2021-07-04 00:02:53.000000000","submitted":"2021-07-04 00:01:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799051-new-release","meta_rev_id":"5aa1ed2850425b33cd7fe3c4734057d73777e655","_number":799051,"virtual_id_number":799051,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-04 00:01:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-07-03 22:26:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-07-01 13:10:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-07-03 22:26:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-01 13:06:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-01 13:10:21.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-07-01 14:17:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-01 18:30:54.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-07-03 22:26:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"b25249835fd7d0d642c4bf925005a1567b120695","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-01 12:45:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8554432a7009e46d289abf52a6c806ccb4ff1afc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 13:06:16.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/566970a1b02343b7a2f7a940c1cddeb9 : FAILURE in 17m 44s","accounts_in_message":[],"_revision_number":1},{"id":"70e6348ccfef3bed558460352075e9b3af280103","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-07-01 13:10:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dc2c196db3aeed837b196704fdd4d056987ffd7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 14:17:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6481908ccd043e9bc9692677657e452 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/862f60430acf4a889bd8684922cbfe0c : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d73f4683629947ffb0e42be5c0231b03 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e13672206864fbdb4317d9c7f519bc5 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4de3c2592f1a41f993b593dc796afb7e : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d2853d2eaf0a48f692f62ac0f1a20d8d : SUCCESS in 4m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a0211126ea5044ceb5fc0226194ce039 : SUCCESS in 4m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/481e7bba79454ffb853467ce4d145610 : SUCCESS in 11m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/67da87b12a82463eba00ad291d9db15a : SUCCESS in 11m 18s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/3ad5cce39fb04ce89d0fd7083bc8a97d : SUCCESS in 5m 37s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5dabe7d9035f40ef8893c8cbcb33ab3a : SUCCESS in 10m 08s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/065f304294574861a343a4459a5d3fe0 : SUCCESS in 6m 34s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a3fe1a29be4c4dfbb19c6b0a1116ee0c : SUCCESS in 9m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/23107552d78447fbbaf4f1c4e0dc294b : SUCCESS in 8m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5e1be9e339b244e4a5004a0a4b924621 : SUCCESS in 9m 27s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/15deb74f29c0498fa42b0c69ddd25d48 : SUCCESS in 15m 57s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/21f1e064e7d94ad4a8026a4f8cf511a0 : FAILURE in 11m 17s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e604f24ddecc4fabadceeda5be1734ad : SUCCESS in 5m 41s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a305713abcbc40898958dde95476cfd6 : SUCCESS in 21m 42s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f49bb85b77114b18b99eb3cef3d14f60 : SUCCESS in 14m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7e2fc28711cd4c638f05b26b729242ea : SUCCESS in 4m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6cf604618dfa464ca0be8375acf38c84 : SUCCESS in 22m 10s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2f5a2e48550a499e9c98e333b6a524b1 : SUCCESS in 6m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/22a42a407eb740299b81cdd922325e89 : SUCCESS in 7m 38s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/929d6902aabb4fa8afdc74d60ee757ed : SUCCESS in 13m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b36b4afdf22428db2d994d460f4c9d6 : SUCCESS in 50m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a2b2b2c9b93d4ed19651965bc5e517c3 : SUCCESS in 1h 25m 55s","accounts_in_message":[],"_revision_number":1},{"id":"91d3e6355ed1d0e259f666fd5b5891ebe768960a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-01 18:30:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d126ad8523701b51385591d4756b7dc681ec93af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 18:51:25.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/345ee56f7f7a431aad6f46316f32b47b : FAILURE in 18m 43s","accounts_in_message":[],"_revision_number":1},{"id":"8cdd61eaf737d0992b0d54cf773fc2b1d4708550","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 19:56:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d15659424aed4fea8c992933bd02c170 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/912d483a3c4f447e988afbc04555250d : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c01bf3227303449c8e55806dd481401a : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2acaab81d924abcb62b91055fccfeb2 : SUCCESS in 6m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8933f0cd1d9a42d8bc5f87625c249feb : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fe3e13b065f746fbbf3c5a9542c69ea6 : SUCCESS in 6m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/28ab1dbb97ee486485d25d013e08ea76 : SUCCESS in 5m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/321fd74c797e48139ee5d91c1bb5af33 : SUCCESS in 13m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dc60706bb74648878c5d12cae861026b : SUCCESS in 14m 35s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/5078c11b8bcf4e1981c1a558771540d9 : SUCCESS in 8m 49s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ffcdc3176340403f9362a8900d756987 : SUCCESS in 11m 57s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/150f78f19d184301a6b991719590aede : SUCCESS in 8m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0ae644a0af0442b7a44eb6cc8f4222b2 : SUCCESS in 10m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9ec0542cd4d74fba93cf474c238fff1c : SUCCESS in 11m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fb40b714ffa142c182393bc618fc5ada : SUCCESS in 10m 16s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/dd29edd6d72f47d192cd488fda9ca3f3 : SUCCESS in 18m 24s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b033d4ee08c24f668e583f8cf4cf0403 : FAILURE in 11m 17s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9d8faa4556e742b5a5425ba9c29a5a3e : SUCCESS in 7m 00s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/fbe79515bcbf443cbd6af8cb71b82a27 : SUCCESS in 21m 45s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/55d78a3063c24f44b5a09ada6b431448 : SUCCESS in 15m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/67b565c1845e4d4891b0d0a8df55890c : SUCCESS in 5m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5c48e74e6afc4768aa7851673b680a32 : SUCCESS in 23m 45s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/58f8185103e44a68b160be3027b56d30 : SUCCESS in 6m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2efc91127968444f8e2c4887111d370b : SUCCESS in 10m 13s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/35b102e67fbe41feb69e82c3da934c09 : SUCCESS in 18m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1514efabe5e54ee6afe24bd8014bb4f3 : SUCCESS in 46m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/003f54c5c3784a658ccf949efdd0e920 : SUCCESS in 1h 20m 57s","accounts_in_message":[],"_revision_number":1},{"id":"893c4feabd60b672104dd8abca38402cdac2b209","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-02 01:21:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"29eaa72246eacb52f210ce62b2806989cfb35cc1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 01:39:58.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/acac2804569345b3bc234e976c639770 : FAILURE in 17m 16s","accounts_in_message":[],"_revision_number":1},{"id":"8eb41244aa09d419ef3d755acd3d5646fc073429","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 02:44:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7133e13b46304bf0938513b0614969a2 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3888f70213ae40ea888e237e54aae56d : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79ea1313044a47b583e942be320a106b : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ec6c8122e3540b8a519db0426016f1e : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3650bf5299c4d8eb20c17aa6b3a310e : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5521dc5f39c24b72ba9cb7ad7a9a09e3 : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/56736228dc0942d4a121bc5d2e926aa2 : SUCCESS in 6m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a65c7d25a6374698adaa2a4ebbd07fc2 : SUCCESS in 14m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/305f65b8473e4b92827d07298cc118d5 : SUCCESS in 14m 31s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/0b2be2c0e8584ef59bf667af206c4780 : SUCCESS in 6m 06s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0c094b9807a04c56abe8eb115d224970 : SUCCESS in 13m 37s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d44c216359b54930b33bf07950546eff : SUCCESS in 8m 00s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a352cf6c61624639a0ab55817419479c : SUCCESS in 10m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d195f21ece3f4946bc64000765863ef3 : SUCCESS in 10m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e02fea0b103640e9961230c9fe625e39 : SUCCESS in 11m 29s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/9d5af44471f84f61b4fa921b77a622c5 : SUCCESS in 15m 08s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5aace47e2aa244509fab4cad8f0c01f9 : SUCCESS in 13m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7c8e9a97a9d44a278309aa4a2e3b6ced : SUCCESS in 8m 08s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c8d9542d68544db98b64fad94c8fd6e3 : SUCCESS in 27m 36s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0f246ca4acb04871b70c908e0991015c : SUCCESS in 12m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5add9db262c444d09a766a384e242879 : SUCCESS in 5m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dc442767aca646ae8e3403bb2dd9274e : FAILURE in 22m 27s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/100b2543d3684edfa066c04f06343a6e : SUCCESS in 7m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/565709b1e184403b997f0d8c06211b62 : SUCCESS in 9m 58s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/21ea786c441b4d66a60316b76b78f657 : SUCCESS in 12m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/303eafadb9da41fbbdbc3aec18d18975 : SUCCESS in 42m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/de941c763fe043f8b3169796b4070d90 : SUCCESS in 1h 19m 41s","accounts_in_message":[],"_revision_number":1},{"id":"7f95e3e25fb45bd3fb108122042312081694e949","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-03 06:58:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1a03f7124fa943f15b096fc6c78c04f63db87689","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 08:05:48.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6354757628274263b5051753609b0604 : FAILURE in 20m 35s","accounts_in_message":[],"_revision_number":1},{"id":"18bf338dfe1111d079f8138b247ae5d82afe6b1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 08:17:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de2585871ca447e2b2843b22aff80d34 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f37f9df240c0439aa471918fc721aeb5 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0c2f8b12b984e35a6cdd9de59858d54 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a316b728c274ef787be684e2e3a2c2e : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d142c18933664ecc9d800d8369f163a1 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/127c2d028bbd4f8f92554977061ff47a : SUCCESS in 5m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/efcc4733d4fb446a9eba65c4c5afda69 : SUCCESS in 4m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b698a8d1accf4ed09f1175ed8b932f4e : SUCCESS in 11m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/17ca8a41021442089ab8fa8361dbfb2f : TIMED_OUT in 40m 59s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/17be11e0a8ef411d89db131d1bc18c99 : SUCCESS in 5m 54s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/fc29aa7ad4e6439f834f5ad9c57a425c : SUCCESS in 11m 35s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/592e416d457244cb8bc7c75a39f275d5 : SUCCESS in 23m 13s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e0b53f2245ed4456a6173f9175f45eb7 : SUCCESS in 9m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4b4c93edeb15444e85d7896663b80881 : SUCCESS in 9m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8fbbc9168563458f9d1231925bf06940 : SUCCESS in 24m 17s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2f7cc38f3ac343128fe3ec4bda92022b : SUCCESS in 44m 11s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/aa468f77dede450899560933af369391 : SUCCESS in 29m 43s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a6f3ffb5901a4b00992e797f8a0165f7 : SUCCESS in 5m 36s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4ab94d44bdac4755aabc38be52361931 : SUCCESS in 24m 46s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1df5d722ef8e463d8689a8ce6f981cd0 : SUCCESS in 37m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bba8cae102ec45649683ee1e17d496c0 : SUCCESS in 5m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4bdbf8f7d14f444cbc8ba3b27d65096d : SUCCESS in 20m 14s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b7590bd76ac54b63b52a7a35fdd983de : SUCCESS in 16m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/10f300b83840436e83188b27421ae3b8 : SUCCESS in 25m 23s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3a5ca1a19ff044c2a949e07e45874892 : SUCCESS in 13m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/75e75193b9064b459a018b717ec5b8ab : SUCCESS in 41m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2ab509dbe7124f129e6a525cc9ce6049 : SUCCESS in 1h 16m 02s","accounts_in_message":[],"_revision_number":1},{"id":"958ba9c5ade08c25102f7782ad519a93b585f667","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-03 19:14:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"818de9f478dabacaae6378a413c29cc039931f04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 19:32:48.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c613ca5ef4f44fcdb077072b46015c33 : FAILURE in 16m 17s","accounts_in_message":[],"_revision_number":1},{"id":"61e526df8bbe82e1e6d7c4b4fa6a82e8ad3e695b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 20:25:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/286d4939f3564d1a9ec13eabb6eb61cf : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/309c39ff89154450a0547742c6b50329 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c19c590b648a4e3483595d899b4a98bf : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f91d9412337946d2bbdab7f7f7f51abc : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d714bf83de2e4e67924fd02a024d1326 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/65b7dbc549e743a4a75be0bc11941706 : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/11a99b17178a426bb0b08780c5384b4f : SUCCESS in 4m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9c3cf8b148ad40fcaf4e861bfbaf36ce : SUCCESS in 10m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9b3982b56eb148ffb1d4b5d3d6b52fca : SUCCESS in 21m 30s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/51fcbebcb14c44fcae4d90eafc68bdfe : SUCCESS in 8m 13s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/1aa0ee18df0d495baea08c47be13705a : SUCCESS in 14m 45s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ecb3af57562c43d9ae093f7f4612a2c3 : SUCCESS in 8m 05s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d6b18781e82b465ca23b3eb187fdccf8 : SUCCESS in 10m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/85acf0c1b9d348a594290a69877a4528 : SUCCESS in 10m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2fa17b7f206a449ca1c89dcb3ae1df5d : SUCCESS in 12m 19s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/639586bb64ae4b3da71a35cc6b3607cd : SUCCESS in 14m 17s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/baeae413295b4b089743f616eb3cb59c : SUCCESS in 14m 15s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c1dbc98c01f449748a70fa3f1f78ee86 : SUCCESS in 7m 42s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/0d15f958a64c4424a6b649813e219fca : SUCCESS in 24m 32s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ef6950a1e6c742d6b8e06bfcfde6a73d : SUCCESS in 15m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b95e1f4475584bf89a7ad2d766edf905 : SUCCESS in 4m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8ad64d7d011c45289198c6ab886d6bd8 : SUCCESS in 22m 18s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/802e7d0b659f4e978053731ece32a689 : SUCCESS in 6m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2bf198f3f8b141c89ba2cb4d3a93f851 : SUCCESS in 8m 47s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c928fa6a6cb44d1bb2674c58946a5734 : SUCCESS in 20m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d48aab81d7554f159854e23d13e38807 : SUCCESS in 49m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9db429d6b2eb4561b2f1aee756f38c66 : FAILURE in 1h 08m 42s","accounts_in_message":[],"_revision_number":1},{"id":"40d8bbd066f481a90b01c74b7c28436336168c80","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-03 20:26:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5c44754bdadaa4227fc8a78f92c3ad28c1989524","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 20:45:38.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d19c445b840e42faadd7fdc9c091f475 : FAILURE in 17m 02s","accounts_in_message":[],"_revision_number":1},{"id":"070dc9585baa7eed160763ff292236f3ccbea8dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 22:07:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e41b63bdb0184b90aea9f82bab25dbee : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1df45c4158514a038b6802d80d66de7f : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5445b532b5a3477e93c6dd01764ecce2 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98c7fa28f3234553b1c7f469bf27ce00 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b574d897e1645039ef47c0dd68f7d77 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/46311c6a746a4e7fbc3275e9b2cda36c : SUCCESS in 4m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d01c47cff63d4cff8fecea2c087af2fe : SUCCESS in 4m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dbb6cbc7555e434fb48a15eccd45948e : SUCCESS in 11m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fc036e027e064bd6a4cf3f796b3af1ec : SUCCESS in 11m 40s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/5689b52027e444f8966b3249d91d42ac : SUCCESS in 5m 51s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/32a40364e733405f8733549dd19940c2 : SUCCESS in 10m 03s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a87df5e2fbf740a981db0f584d9bd984 : SUCCESS in 6m 27s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ff9de34a321e4ddf97b48c276df937ac : SUCCESS in 9m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3cfd06f9ed4341a5b5c49b581fdf6cca : SUCCESS in 8m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/260a31896c754b14873de46de26ea82b : SUCCESS in 8m 46s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f33e9d128b4c460aae052909bdb15666 : SUCCESS in 19m 14s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c2e4857cdef74680849b22f3232c1f1b : SUCCESS in 11m 13s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6c93f4a21fc44a30bfcb49c896a3d38e : SUCCESS in 5m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/64e3875088704c9982718497dcbfa5af : SUCCESS in 22m 26s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/87109c066d004c758d66827b76e02fbb : SUCCESS in 12m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/aea803ee6ed340ea88065dabfaebc796 : SUCCESS in 4m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/47d5f3a35655400eb9d72625f52283e1 : SUCCESS in 17m 21s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d210b4c998ca4bb396bc7d990dcddc66 : SUCCESS in 5m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3b65f82024f84825bc91d0528f3c505e : SUCCESS in 6m 40s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0d3e0717c3bd4d15b62d5ac3fa55b2ab : SUCCESS in 13m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/436dffa659e242e280570c204c7984f3 : SUCCESS in 47m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cfd347d2b8184396b64411d1f137a1c9 : SUCCESS in 1h 38m 08s","accounts_in_message":[],"_revision_number":1},{"id":"2a2bc9f97353c6124cbb3f2b554a8a65f64f2f01","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-03 22:26:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d91f9cb66d8c255feed8785ff4872caeb077caef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 22:27:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5a9af0c73f3ceccbf5ee675a43f649dab839dbfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-04 00:01:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db475e64711b4a60959e86a1e4394f39 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdcfd818c4c1480f8dee1a71b07714d7 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c950b640fd04eb7bb3faa2bf48235e4 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/109a0595eaba4a35859eb4ee8817f9fc : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/70790a47f2b54e67977f9ec61cb1c4d6 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/da781644b1b74c388b30f6187763b033 : SUCCESS in 3m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cc750794b57243b5aeb75351af9ab260 : SUCCESS in 10m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/113a1c13403744a2a9164116c774b07d : SUCCESS in 10m 19s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4f9160d5156a4054a20f3664cfa09e1f : SUCCESS in 9m 53s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9ef090db547c428a89e98d2cdfad13a1 : SUCCESS in 6m 13s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/783b26df8246452ea131bd9995b50a78 : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/47323c34e7c449a1b8301579c136a5ba : SUCCESS in 8m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5cbbdeb1d7824e1099018e74f31d9018 : SUCCESS in 8m 53s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4ebc07f8157243c99983990a118f35e9 : SUCCESS in 14m 43s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/445143840d764f54b2d8132d5eb508fe : SUCCESS in 9m 18s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2fb46d6f5feb40f98326f64ac385cfd2 : SUCCESS in 5m 23s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/de4443644cd44b3d8ef4a0b1400ad83a : SUCCESS in 25m 31s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5a4cf8a252774b0183677a25fbed0499 : SUCCESS in 12m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7975d00a17334ebe8945fbff49eb70a9 : SUCCESS in 4m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9c07bab7302d40a3b3355f78c7fa6a0d : SUCCESS in 17m 41s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/76332fbc80714566b4ae405ce1e40b93 : SUCCESS in 7m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b5dd4a19b3bc4b0f8689d1355ace0955 : SUCCESS in 6m 33s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/49d12ee1789a480d92029612059a7b09 : SUCCESS in 14m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5a5af765d094fa8979b3d6f11fbf48e : SUCCESS in 40m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6091d8ea36e14bdc93716326ac9c5724 : SUCCESS in 1h 28m 53s","accounts_in_message":[],"_revision_number":1},{"id":"ac2a3e0e2ad13790184e51bc62bb8b2574c04a14","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-04 00:01:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5aa1ed2850425b33cd7fe3c4734057d73777e655","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-04 00:02:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e467708e04f14bee9d5a10d321f72809 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"17df2648a2c3225cb9af4d6784824eaa8c594e8b","revisions":{"17df2648a2c3225cb9af4d6784824eaa8c594e8b":{"kind":"REWORK","_number":1,"created":"2021-07-01 12:45:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/51/799051/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/799051/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/799051/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/799051/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/799051/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/799051/1"}}},"commit":{"parents":[{"commit":"5c4901e38e5ee51cd28cf6e42ca904f19a001f47","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c4901e38e5ee51cd28cf6e42ca904f19a001f47"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-01 12:45:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-01 12:45:48.000000000","tz":0},"subject":"update constraint for oslo.rootwrap to new release 6.3.0","message":"update constraint for oslo.rootwrap to new release 6.3.0\n\nmeta: version: 6.3.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ie08cda2ed6285c9ce506432cb516161a85b49076\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ic55c0d8912e3934c4dab5bdce3d714a327afb081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/17df2648a2c3225cb9af4d6784824eaa8c594e8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/17df2648a2c3225cb9af4d6784824eaa8c594e8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
