)]}'
{"id":"openstack%2Frequirements~799052","triplet_id":"openstack%2Frequirements~master~Ibf1854cc186e0416edc6042ecad2d4ff03712fcf","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ibf1854cc186e0416edc6042ecad2d4ff03712fcf","subject":"update constraint for taskflow to new release 4.6.1","status":"MERGED","created":"2021-07-01 12:46:38.000000000","updated":"2021-07-03 19:08:46.000000000","submitted":"2021-07-03 19:08:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799052-new-release","meta_rev_id":"adbe701c5c9c06ff4b8baedf1c60470f20a1f8d4","_number":799052,"virtual_id_number":799052,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-03 19:08:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-07-03 17:48:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-07-01 13:10:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-07-03 17:48:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-01 13:10:39.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-07-01 13:11:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-01 14:09:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-01 18:30:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-07-03 17:48:03.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"12bd3b45233a2edf25379b59e160817fc1c8220c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-01 12:46:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93485a69808a4c587987c90a53c4a05be460dbb2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-07-01 13:10:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f05f22cb0abda4294f7851fb5bce1514d49a95c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 13:11:44.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c621cdedac984c2ca3e506a02e588938 : FAILURE in 17m 05s","accounts_in_message":[],"_revision_number":1},{"id":"fcd8bfcddf27ebd599e0158b6cb34345dcf926c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 14:09:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76e58c2c6e824d63b5e1fb1e799fcb30 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43057f1952754d328e52cb86d32f9515 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df6712b896d844ad8c1042590b0f9d25 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bcd5b2b07986421c89e415da1eb8adce : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37c3ef85836640e8a69adac54aa7dd6f : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7dfe1e2ac4c7441b80744d30adade4c7 : SUCCESS in 5m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9ecf7d11a80d49f19f45f568715facac : SUCCESS in 4m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a1e0b8838a754490819e0131ce5796c9 : SUCCESS in 12m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4e8ca600941e4520ac0e8a7521c8604d : SUCCESS in 13m 39s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f92ad3a294514ccda4a694781f22377d : SUCCESS in 6m 20s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/520219d6501b4631b722a67610aae612 : SUCCESS in 11m 00s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7a1078b9e95142a0a8dd8fdc44147679 : SUCCESS in 6m 28s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/28be06eed8cc4bbd8656a99a00b31575 : SUCCESS in 12m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/96ab7e1b89db4f4a81eefacb154792ab : SUCCESS in 8m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d383d1626b8a4d61a1a705de3ce77360 : SUCCESS in 10m 55s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a5da88e753064db68a4aef83725a791c : SUCCESS in 17m 39s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6f16e98562e044c79d14c18509552afc : FAILURE in 11m 20s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/424ab4e03f1f4d0ca68c30712667d481 : SUCCESS in 5m 40s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/5c0d6a3250d5430b93c74391fc76b9c0 : SUCCESS in 28m 50s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6221ec5ca9754ed8938639e945b40bd6 : SUCCESS in 14m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/34fcd28bd6ae4aada2b19b7d69944c0c : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1bd3b1d9dd4a4312912d61d77bd033af : SUCCESS in 20m 08s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f2fbdad7b4514261a2a592c3fe5a3aef : SUCCESS in 6m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d1b00a01624e4b809f3cf87983a48d0f : SUCCESS in 7m 53s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0cf4ad59e134472b9b13a4adf271dc09 : SUCCESS in 13m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ab2fc201a3e46a092b6ec3f8c411325 : SUCCESS in 56m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f8521a101dd44c1da83e2ee781532ee4 : SUCCESS in 1h 17m 48s","accounts_in_message":[],"_revision_number":1},{"id":"aeb97297acd81bacc74e5af2f606fb1be527b474","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-01 18:30:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5c308401288b759aa2427792d5a02c97c4996eba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 18:52:21.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6cbdac428f3d409284035c9eb3aba676 : FAILURE in 19m 18s","accounts_in_message":[],"_revision_number":1},{"id":"6a3594bd8cbad9a10cb646340eaabfa525f33ce2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 19:52:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ffdd6019c4e4e2ab4e80a473770f448 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d932168b196c4babbbf0beeea9eaefec : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19e308e4f3704928a16b61e16e987268 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eaf768c7f8fb4e7880e918606a3775e0 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c7f0baa00c140059aed0d28922e4869 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/920be96f990d4c34afde46f99fa25e85 : SUCCESS in 5m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4a82b57ef85a4db9aae333072bd15982 : SUCCESS in 6m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bea44c01a01f41d588c224ba4abdb1a2 : SUCCESS in 11m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e7e595f549e34efab67f857d68918fa9 : SUCCESS in 10m 34s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/130048e1f17345ccbe049e760ee53cf0 : SUCCESS in 8m 42s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f5973e92139049df9099b2a6e0d9e8b9 : SUCCESS in 11m 45s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0a2f295969de4a2cbea1f4b4e4525dce : SUCCESS in 7m 58s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6d9398ca34064f25b763339e38089287 : SUCCESS in 16m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5de0a79fe1ab461abfda75c340e240d0 : SUCCESS in 10m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1852e597b49e40c8b89ff8663b5e3186 : SUCCESS in 11m 05s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5fabfc24a61547a48dd011748e83fc59 : SUCCESS in 16m 55s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/bf63f666210f440ea0c00a3f255c2160 : FAILURE in 10m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/385c9475b09a484e94e9401f2f638475 : SUCCESS in 6m 39s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/fee57841e57b42d79d594d5774af5faa : SUCCESS in 26m 29s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0659d37558364640b1e170577fd4a3fa : SUCCESS in 12m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/900b1a86899e461887868bab6f6bb096 : SUCCESS in 6m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1e0f03e3b68d40158c50af716638c347 : SUCCESS in 28m 29s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d1f7fe658b6b4cd1833b243474120f87 : SUCCESS in 7m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e897c866494d4b1db0611f34a78fa23d : SUCCESS in 8m 45s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/257ad2a6f7e4474792571c11861a704b : SUCCESS in 14m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c6748c1b0a9c40769e43e57c82a85fa8 : SUCCESS in 52m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7396e30b61c44ce2be4293e34f4cacb8 : SUCCESS in 1h 18m 23s","accounts_in_message":[],"_revision_number":1},{"id":"a8f6588eb2ec98fc1bda5d26dd7d73fdef56fc59","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-02 01:21:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b582cc223ae3dd11b9b66c239acc879917a49ae5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 01:39:52.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2ffbc814fe4c43bea2a587ddc874bb12 : FAILURE in 17m 10s","accounts_in_message":[],"_revision_number":1},{"id":"160e99fdecf09587152d76a5697d59883c06e458","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 02:53:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/092aba179dc844fbbbad171222d91641 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23db24b0d3544ba686ae97def7f810d0 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db4ae70d255047759d775477afeab428 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/852b4e2b66ff4e07ae848c4fc2f3a034 : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49bf0e3a916f48b39788f7290b83855c : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bc5bfc9229f140cd81cbbb676c5eb034 : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3050e54df58948708f2f9d2001e5ca22 : SUCCESS in 4m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f49f78c61f7e4e94b22dc4d26b322866 : SUCCESS in 12m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/813720eb43c34892abaacbb1430b7261 : SUCCESS in 12m 21s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/72887f6ae40347b3ae857c50b9eba4c1 : SUCCESS in 7m 23s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5e9fc57834e346b385ec4d7d33d0d79b : SUCCESS in 11m 13s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/1745734815624aa393bb9261e250608e : SUCCESS in 7m 25s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4ea2cae75b1b4187aef8015dcb9972ac : SUCCESS in 11m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/796aa3d01671415abc6dcdaf94000a35 : SUCCESS in 9m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/eb62c0282b084b469faa2f7e83ed95fe : SUCCESS in 9m 37s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/16f1ad2d42ea4b81918feb0cfe28143d : SUCCESS in 13m 37s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9594d286b0ab45539edf08f3273bd356 : SUCCESS in 10m 07s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0f4cdd8ae0dc485081aeba7ab6a3fdf7 : SUCCESS in 6m 08s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e64a3c607ca14311a68467857ec2ed04 : SUCCESS in 28m 37s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9b4784c32ad344c7ae1b724cc4cff1af : SUCCESS in 14m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bb388260db914eb0b7360186d88e34fd : SUCCESS in 4m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5c94f0ea395f4526bd18f0342de0dce4 : FAILURE in 23m 52s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2a089ea8f8d84346a08e9b7f280d632b : SUCCESS in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5febb6cf0d414bdeb74ea8f55d6c3370 : SUCCESS in 8m 14s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5a252176953a41d29e7ea360009314b9 : SUCCESS in 13m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f0a51bdbd6142a99f4be045e6b23da5 : SUCCESS in 48m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8721685088984d329ba78d29421f472f : SUCCESS in 1h 25m 57s","accounts_in_message":[],"_revision_number":1},{"id":"04e15980733b9f095b57d3c4c52cc96b3297a671","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-03 06:58:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d28904c4b9f9b9312011b247d50c756e2ffab467","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 08:18:24.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/686d8503873c4b96a00cd8da1628999b : FAILURE in 23m 10s","accounts_in_message":[],"_revision_number":1},{"id":"ae6c5503d026aca5f4088248be554711ee701887","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 08:39:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f14ace4f2834d1f9d295594fca52ca5 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/110e704e033c4fa387cd648389de0d83 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2afb6b704c54d148f6a7b3b062483b8 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c795d6d75b44b7f9d9fc40818077fd4 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9ddb56727a64521823f0d186b3968c6 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d091ef9ee8464a80bf9d54e64a595e9c : SUCCESS in 5m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/071693309af9433c830104c7d6cd703f : SUCCESS in 4m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/13ff3b98fb9340f9a4000c3ceb323a8f : SUCCESS in 12m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/39960d520f6a4f8ea5a584bb8c74c332 : SUCCESS in 10m 19s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b0b223d6816e4eda9116f4b3014be794 : SUCCESS in 6m 40s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a6cd320c6c054029adfeddbed8f65f75 : SUCCESS in 9m 38s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/315d721895ef442ba1ef745a5f297944 : SUCCESS in 7m 15s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f8cc53215c604be797d3415a8ae29bbe : SUCCESS in 9m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/37c8d6b6da4e41b181406e1df8c42487 : SUCCESS in 9m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d20ec50fef9547af9e1fd7d9ab608b8b : SUCCESS in 9m 33s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/819bbccb34364da9a491ec444a0a828b : SUCCESS in 13m 01s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/23b197884fda4469887d503e260f8187 : SUCCESS in 30m 32s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b4264c4d20cd4f2386b5ec4b440b2ccd : SUCCESS in 6m 57s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/57a92cb0d4a64244b23d0d7bd2ae1755 : SUCCESS in 25m 32s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3fb942b5ecbd466dbd5c843beaaa5561 : SUCCESS in 14m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bab329d4f49a401292c789766158a37e : SUCCESS in 9m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a44b599f56484f3a9686d7e3def584fb : SUCCESS in 20m 11s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/95bdc8a9c0424bd28bcf317892424451 : SUCCESS in 6m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/30d900fa1b034c8d937d2548b7dd4531 : SUCCESS in 6m 44s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/008726e38a9e4ceb9dd0933d6ec4aaaa : SUCCESS in 12m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/96ad5d83d1614f518ddbdb0fe4720773 : SUCCESS in 1h 27m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f201325d6119447cbe29f7ec90da82a9 : SUCCESS in 1h 18m 13s","accounts_in_message":[],"_revision_number":1},{"id":"07aab2ed60bdc699cde390e65723d7e92f2ec51c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-03 17:48:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"569b28550b09a654d9d3a06ff3f10f1b6a7e445b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 17:48:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c92766b1bc2e6b608956bc06cdb46669802eba57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 19:08:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/215c887878314708912dc1551e203151 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94d098e5de1941d7be58d9fc1f049865 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afe19751a55f43629a7fb84646e2cf4b : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30dab737b95f4673abe846059a7b6096 : SUCCESS in 5m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b21cc4205e96402a811a68bc00e82798 : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d393d49b1bb04558a0c1ed9b59141ec5 : SUCCESS in 4m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0b0b973583c94290aefca1cea5529398 : SUCCESS in 11m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9d5c4c6827e741d8a9b060fe9909cb1c : SUCCESS in 10m 41s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2b25a5146cca49b293f1ab109b464c54 : SUCCESS in 10m 31s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c2bd73f743674851acbbc5ae51315f9a : SUCCESS in 6m 15s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f94c98926fa8492d82e1bc6c4ab47064 : SUCCESS in 10m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8c254f93a8354cc486b5dd12c087656c : SUCCESS in 8m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a1ae00d8f9fb499598bf10361160ceaa : SUCCESS in 9m 00s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c2a20521d0964945bcc616620263adee : SUCCESS in 18m 54s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/502d9e2b997f488392a1c2a0611ca878 : SUCCESS in 11m 21s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/113105cb32954da4a92682c69dc867f0 : SUCCESS in 5m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a9359e2e87234d3bb10eccca179cd102 : SUCCESS in 22m 22s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/df6b1f7e65834a2086614c3b8a42a1a1 : SUCCESS in 12m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1624b76d1fb04d80aa0c7263a12fab1e : SUCCESS in 5m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/42499b16d8e9467e863bdfc6a25c7b44 : SUCCESS in 18m 06s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1d5ffdb4a5b0419f91ff5b6b52d21c3f : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/37966c09792c444691264f8b317e6440 : SUCCESS in 7m 22s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6f74f6ec0952418a8d3f6a3964dc7037 : SUCCESS in 14m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/839edeba9f524fc58581d7a7c729b6da : SUCCESS in 43m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f0b07b33f9c44f31857727c85a78a6cc : SUCCESS in 1h 12m 30s","accounts_in_message":[],"_revision_number":1},{"id":"adbe701c5c9c06ff4b8baedf1c60470f20a1f8d4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 19:08:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"83a36b180c40682c217cb3fa541c87428b87ea18","revisions":{"83a36b180c40682c217cb3fa541c87428b87ea18":{"kind":"REWORK","_number":1,"created":"2021-07-01 12:46:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/799052/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/799052/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/799052/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/799052/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/799052/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/799052/1"}}},"commit":{"parents":[{"commit":"5c4901e38e5ee51cd28cf6e42ca904f19a001f47","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c4901e38e5ee51cd28cf6e42ca904f19a001f47"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-01 12:46:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-01 12:46:36.000000000","tz":0},"subject":"update constraint for taskflow to new release 4.6.1","message":"update constraint for taskflow to new release 4.6.1\n\nmeta: version: 4.6.1\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ie08cda2ed6285c9ce506432cb516161a85b49076\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ibf1854cc186e0416edc6042ecad2d4ff03712fcf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/83a36b180c40682c217cb3fa541c87428b87ea18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/83a36b180c40682c217cb3fa541c87428b87ea18"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
