)]}'
{"id":"openstack%2Frequirements~799053","triplet_id":"openstack%2Frequirements~master~I10c93abfcaf5aed5c606330c9ef3a19c5a099d40","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I10c93abfcaf5aed5c606330c9ef3a19c5a099d40","subject":"update constraint for etcd3gw to new release 1.0.0","status":"MERGED","created":"2021-07-01 12:47:58.000000000","updated":"2021-07-03 19:08:50.000000000","submitted":"2021-07-03 19:08:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799053-new-release","meta_rev_id":"d758b636e37e9b1465cb53240ad326c5f05fb5af","_number":799053,"virtual_id_number":799053,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-03 19:08:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-07-03 17:48:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-07-01 13:10:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-07-03 17:48:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-01 13:10:56.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-07-01 13:11:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-01 14:02:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-01 18:30:58.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-07-03 17:48:04.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"92f738b1c686f3f09cb2dd061820e274706a6b40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-01 12:47:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"287a2fb7f546207e24a7937f054bd6e2f9701b89","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-07-01 13:10:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"268f1ddde7a136066cba30c69fbcb60c7dd453b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 13:11:21.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/de867137697f428e952c9927a6a7563c : FAILURE in 16m 32s","accounts_in_message":[],"_revision_number":1},{"id":"92a3da3c963324a08bbb4deda36762e44dfd6a81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 14:02:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f74891befdb446a9e983fca4f0c5a00 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b5972e7e07c43a7bbfdd445bd0168df : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e086c868a394a67968205013505016b : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3603d1ccab64dba8971cc313b3a2877 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6cd7049129374159aea6a6c4d6137249 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a874bfd12d9841e79132ddf5eecc2000 : SUCCESS in 5m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d692aef391e240ce8644404252c8af6b : SUCCESS in 3m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7d4a99969d5d4bb48b6338426ccf3cd9 : SUCCESS in 12m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bfd8de6c514c4f0795200530d5483257 : SUCCESS in 12m 17s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/c1a75281cad94cd6b2c95e866aba654c : SUCCESS in 6m 07s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/cee3d646ba0c4d6697ec316b373c6689 : SUCCESS in 10m 37s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/113a8a40959d4ab8bd64d0909486abf1 : SUCCESS in 6m 48s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3a87f7d98e71471785d097923e18325d : SUCCESS in 11m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fdb12132b8194147a80aa1141163768f : SUCCESS in 9m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/15adb38e730a460e8faac347bb64f599 : SUCCESS in 12m 04s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5c953c200d5e4739bd86cf3416c4ddcf : SUCCESS in 16m 08s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9c79e1da5e5741f3bdcadd400a610fd1 : FAILURE in 10m 13s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7ca1e862e19a4f7d83d4b0c710e15422 : SUCCESS in 7m 10s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/133b2a2a810f46669822a289ac782560 : SUCCESS in 30m 04s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/034dc06c3bfc4511af8cc63410ce0671 : SUCCESS in 13m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9f0621a84a674b1ca06f5ce7ce2ae520 : SUCCESS in 4m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a6f0cd3c034c46439a8d9502a28dc4f8 : SUCCESS in 21m 02s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2b164197d29f4915b03ddc930b9a6280 : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/49eb54af6b8b4b77994ff7f5f1fcbfb2 : SUCCESS in 7m 19s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/271017e61f104c59b4efdb76cd0b4919 : SUCCESS in 14m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/15331d3fe2864b8cb44bf8dfbaf73102 : SUCCESS in 55m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4697d24c72cb491f9e5d26a495e5b382 : SUCCESS in 1h 12m 39s","accounts_in_message":[],"_revision_number":1},{"id":"eba55f26b59bc25a7fac07c5221762eae821a55a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-01 18:30:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4dddbf32aac8177dc7750d06ee380d62a270d748","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 18:52:04.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/70f6487442c6411ea01d86d45f8ea146 : FAILURE in 19m 12s","accounts_in_message":[],"_revision_number":1},{"id":"518ba51ce018ce7e3e2f8ca60985015874b516c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 19:53:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdc4421b08244e1980708aa17eec8315 : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/541200d9cb7f4642b6987244c73801bf : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79c4f6847e8c4c17a2fb6863d09d979c : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/526f97c355b242b09637a36daa14b55d : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd9eb4aadd81437086c923a018e4c185 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ffd0860e212641ec96d0911f5078a4a8 : SUCCESS in 5m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bece366cdb6b4a798c94df3d723f79ad : SUCCESS in 5m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/745f0710d25b469786bad6941436eaab : SUCCESS in 12m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/253b171d52434e53814cf7b810255c20 : SUCCESS in 12m 47s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/09cec2d6ac8342a2b4001a964a035927 : SUCCESS in 8m 24s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6e1e2190952d42aea2d12033dc34a049 : SUCCESS in 11m 35s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/68d173c6dfbc4064bf610d1e1b5ff2be : SUCCESS in 9m 41s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f439e10017c24d208a9a2f4c3abd0413 : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8d778a4b9ef8483e84d0db168a9b53ca : SUCCESS in 11m 23s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d8f30159975841baa23eca5b60a5cd04 : SUCCESS in 12m 07s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/138eef02ea964443956dbc37f59747c0 : SUCCESS in 19m 20s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/496fe8e46abb46058d6e5b1cc4cda39d : FAILURE in 10m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b5950b2ae99f4e698dca12e6b05f5b33 : SUCCESS in 8m 17s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4ad32ae3f24f4125b0fea0ea889fbc0e : SUCCESS in 23m 51s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6e435912a82141e0913ac0dc6df712ac : SUCCESS in 15m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b6bdbc1d089c4b8a82eb39848f0de36b : SUCCESS in 6m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9f03780a9f5241c2a7f3480062db7dbf : SUCCESS in 20m 25s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4db1a1ad07924a1ea3ac17ce17492bf7 : SUCCESS in 7m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/136f9e1813b042f7a6e76c9eed2b8087 : SUCCESS in 10m 29s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/84c5055162ad431aa81dec61497d86d7 : SUCCESS in 15m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/35829fa128f04d06955c8374386a83e5 : SUCCESS in 39m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0397f341220342c8901b9aef72d67e21 : SUCCESS in 1h 17m 17s","accounts_in_message":[],"_revision_number":1},{"id":"1151421dfa7ed2c0040a8569d064d8ef0ff23e27","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-02 01:21:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"12732127da6ca440f78bfbab0d2cf57b505b521c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 01:39:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/335d81d842ac4a81bf5225c9f382e907 : FAILURE in 16m 46s","accounts_in_message":[],"_revision_number":1},{"id":"f7371607666928efcc38dcd34a35d15e8742a764","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 02:59:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9bc21ccce5c4cf8967dfc9bf32f6997 : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/698cdecd016d4456aa67b62b21f6607f : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72811dc168684717a316aadd9dcc3f66 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/625661d506c044e2a293cd23b361a3cf : SUCCESS in 7m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e01332e717ab4c31b481625a22efb254 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/df37f9dbd4b648f98015aba186765dea : SUCCESS in 6m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7ab5fcdcbe47451fb7bebbc7e8d114ca : SUCCESS in 4m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/66a966f8487c4bb5a9771e9e5bccb4de : SUCCESS in 9m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c3b58a33027f41d4a480767ed735e7c2 : SUCCESS in 11m 37s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/cdd1a1906dc4470d877308f67285fc16 : SUCCESS in 7m 20s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/33886ac2e0274d1ea120ca2fef4b9612 : SUCCESS in 10m 30s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/02541fb0aea34f9eaec0dc0d40f97e2c : SUCCESS in 6m 06s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6fe839d183b24258b304387406bffb95 : SUCCESS in 8m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b290b4da858345d7a7040faf572a50fe : SUCCESS in 11m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7cc14549b0e34e58ac77169147e76ac4 : SUCCESS in 12m 51s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4a0fb883d3bf4f0186e9a68355f17cd5 : SUCCESS in 26m 31s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/71564ca471624fadbc2fa3ece6986e45 : SUCCESS in 13m 24s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1da3259adc344d27bd735f38c41d5f06 : SUCCESS in 7m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4db779091faf4319806984618ef4ca03 : SUCCESS in 21m 25s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/04a249c016604a01a7384cb2bbb56a17 : SUCCESS in 13m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d61f51256532410392a8e685bbf9f8e0 : SUCCESS in 5m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/86be0ab3001343cba81a58d505f27552 : FAILURE in 18m 53s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/ef414df04d324a918e744fb614994c0f : SUCCESS in 7m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/beab8d5f5c8d4a60bd23f5886b848ee6 : SUCCESS in 7m 05s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6852dfeacb9b49a2831e86578534a43e : SUCCESS in 11m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47560c0d0a31407ba5210982784e4a18 : SUCCESS in 51m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aa265d8aa0424c6cab060d9958357380 : SUCCESS in 1h 36m 23s","accounts_in_message":[],"_revision_number":1},{"id":"0d52476bacdb9983086fd181fcddf2a357038d37","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-03 06:58:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0f328abb2916adaadf32c06c6232ac5a5176d7be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 08:17:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1e853e9a04c94f5b92fc248ad82f6430 : FAILURE in 20m 14s","accounts_in_message":[],"_revision_number":1},{"id":"a3a6d1d8db0ceb7f2780b23749d89f9774128691","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 08:26:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/109fcc54d5c44a9fb7c2819cc3fd48f5 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b090a93d7a447fba2f6246acc2796b5 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58df07301c07446294f3db6336e65a80 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43fcc7a670c04257b85f33073a976853 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9df4107341c84039812a5998122d025d : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b72a558bb75c43d584469ab98ef1d50c : SUCCESS in 5m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/84882a17a7a34b8da5e3dee0b9e1701c : SUCCESS in 4m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a9e33fd4c18442dba54bdfb73b5f957b : SUCCESS in 10m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/145f8e78a85148d5a9a322fd14b2572c : SUCCESS in 10m 06s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/67f859629b954781b6af710500633d41 : SUCCESS in 5m 55s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9d93eafc53d5419ea6f86e332039a41d : SUCCESS in 10m 14s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/1c49d5cb643e4697bd6609d721f9791e : SUCCESS in 6m 42s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/20fd56e3855e438695715422d7b835cc : SUCCESS in 8m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f32c207244cf457cb25a37ebd738364a : SUCCESS in 33m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5be5ab76d01b4347b3e031388254ac54 : SUCCESS in 9m 24s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d525ece1a1874908a3ffcccfae4ccc63 : SUCCESS in 16m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/39a9958e57ae4ad49e085549ff911288 : SUCCESS in 9m 20s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b97b6bc661a944e9a221f9b98a2dcfb1 : SUCCESS in 20m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/81a74201a201488cb05c607d6aaa7ce4 : SUCCESS in 22m 25s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/94192334c66f40c9bfe1bf11c4fdbb04 : SUCCESS in 11m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3b31666cfd6548a59267f27308aebecb : SUCCESS in 6m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3440eed514a249cab3bd4ee16308a902 : SUCCESS in 21m 20s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b35274cca77a4149a617cf24fdf03478 : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8290d6d13299456798c90170594da3a9 : SUCCESS in 26m 15s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3f61bc2085374a0cbb8f7bf48142ca12 : SUCCESS in 12m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/961e00f867ec404ea5908e0b707ac12e : SUCCESS in 38m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/275a901366f44ae7a49b247b9bd65e2b : SUCCESS in 1h 22m 08s","accounts_in_message":[],"_revision_number":1},{"id":"96d8c90dbe15dc0c1a883209d5567b4c8b1bad7f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-03 17:48:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"49ed79d1120ed46680feab5ec871629dd15f9489","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 17:48:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e43a3172a84a5d313c1d2d4a20c5238b1b45ad18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 19:08:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b90b35edcdd4edd968e0e9ec9784ccd : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2249f6cda9bf49e1850593d6701ea2ea : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/512e3ce52a50492d969e57b86e646781 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ceb88d64917046ed89ea3d0cb4eae9e0 : SUCCESS in 5m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/542db9ed788846e0810804f2b484136e : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eb9bb0b01dfb468ebd4fdd76a26ea409 : SUCCESS in 5m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/56246e45d6344cac821547f6ee0e93a9 : SUCCESS in 11m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3ee831997a8d45a7925edcfca9675f97 : SUCCESS in 11m 06s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4be321af241146e9b6cce24f61fe39e5 : SUCCESS in 9m 38s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7156a89828bb4019a614fc21823fc6e3 : SUCCESS in 6m 19s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/02b4e79ac70c4c3f952f0c03c24ff34e : SUCCESS in 8m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3a013e344a2f4de691fbe8fd5e6e8258 : SUCCESS in 8m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2fee9fd03e90450082ad3f867284197c : SUCCESS in 9m 02s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1a2199c4702f4d7386747f0ac460c856 : SUCCESS in 14m 55s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1d7e4be14e104c749d8a4dc57f247682 : SUCCESS in 10m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f2636afc03984b2887d42ba6f6757c63 : SUCCESS in 5m 51s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/355bdc547c2f449394349ccb5c4c2ebe : SUCCESS in 27m 40s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1a9e1cdf9e024f758dc1fa9be9049f87 : SUCCESS in 12m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e1c89dc6e0834427aa461648536e5004 : SUCCESS in 4m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6f452988fa76415dbdf2460a690074f2 : SUCCESS in 23m 38s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/08700c1bdc274778a9658fa32a0e0ff8 : SUCCESS in 6m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/93c0a528ea4a4c6eabeb69fa31b18b6c : SUCCESS in 7m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5a119662f36e4094b82af8eea00807e1 : SUCCESS in 12m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/672dd7b9cbed4950bf1dc27fca063b61 : SUCCESS in 44m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4ef5d98476e846e0b68f89c960703d8b : SUCCESS in 1h 06m 54s","accounts_in_message":[],"_revision_number":1},{"id":"d758b636e37e9b1465cb53240ad326c5f05fb5af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 19:08:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"da0959d4e3abfb980bf2ef800012ecbea6df8de8","revisions":{"da0959d4e3abfb980bf2ef800012ecbea6df8de8":{"kind":"REWORK","_number":1,"created":"2021-07-01 12:47:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/53/799053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/799053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/799053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/799053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/799053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/799053/1"}}},"commit":{"parents":[{"commit":"5c4901e38e5ee51cd28cf6e42ca904f19a001f47","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c4901e38e5ee51cd28cf6e42ca904f19a001f47"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-01 12:47:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-01 12:47:56.000000000","tz":0},"subject":"update constraint for etcd3gw to new release 1.0.0","message":"update constraint for etcd3gw to new release 1.0.0\n\nmeta: version: 1.0.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ie08cda2ed6285c9ce506432cb516161a85b49076\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I10c93abfcaf5aed5c606330c9ef3a19c5a099d40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/da0959d4e3abfb980bf2ef800012ecbea6df8de8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/da0959d4e3abfb980bf2ef800012ecbea6df8de8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
