)]}'
{"id":"openstack%2Frequirements~799084","triplet_id":"openstack%2Frequirements~master~I5b30529be4f6cd732570a7b3a0446894c42387a9","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I5b30529be4f6cd732570a7b3a0446894c42387a9","subject":"Bump osc-placement version","status":"ABANDONED","created":"2021-07-01 13:38:33.000000000","updated":"2021-07-03 19:15:37.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"be6b6fdb57f66b9c92635fe15180a3343420fcdc","_number":799084,"virtual_id_number":799084,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-07-03 08:37:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-01 14:01:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-01 15:06:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-01 18:31:01.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"}],"messages":[{"id":"dba134287bf573f6aefc0503221fcb3f2263bf13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-07-01 13:38:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e32d936123a10e97f64a23cfbba14ee60f239ccd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 14:01:56.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5e4ce5dedc604761939977e147084516 : FAILURE in 21m 17s","accounts_in_message":[],"_revision_number":1},{"id":"3107b4d46dea06f2af8f136bb1558e19e372f0e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 15:06:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef12320bcdf94808a97389dd6fd5b840 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e64216b2662c4a4c95006582edc3420d : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4408a8f2314341529d503542ada646b8 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/142c3c2a70704b3c80768ebe65f6b1bf : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2e4d20194ca4bfeaf52fe08e3b9d089 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7a7992e242a34b39b5612d015958da04 : SUCCESS in 4m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/88ce937c305f4b5281fa4da1211c534b : SUCCESS in 4m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/93755874469e487a849f91436fef8cf9 : SUCCESS in 10m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fc235308f7eb48a68666cf772d1dc604 : SUCCESS in 11m 39s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/22702004fa69452699d27be4f3e959dc : SUCCESS in 6m 49s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b8b0597fe53a4a62bf7868cc151aa36f : SUCCESS in 12m 29s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/fe76260acf7f47f4a5380e693c1d889a : SUCCESS in 6m 38s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8bd3d6ada294436f88798d5552126364 : SUCCESS in 11m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/95ec502c8e384079a06e870515343a6f : SUCCESS in 10m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/17b6aead61bd4c06b91856854e616b7a : SUCCESS in 10m 29s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/08dbc32de00943da8fb25b05c2ab5631 : SUCCESS in 15m 15s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/54a5016fd59a495c85add0234252c3c9 : FAILURE in 10m 56s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f6eab19a67a74333b75e06aa9d6eafcf : SUCCESS in 7m 21s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a69fd3a2a936403e979fa240a9f80d30 : SUCCESS in 27m 38s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/53d3aa242e7e4732bac89f4e69f80f4a : SUCCESS in 15m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/db1e4437d2254c10bfdbd5938092a92f : SUCCESS in 4m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cd980d6401234fa4b2f08bd96bf0e556 : SUCCESS in 20m 33s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/21d05cc37bcb41729842fb95fd77c5fe : SUCCESS in 6m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/74a0b3148f4243aab265f0b3d6c0ad33 : SUCCESS in 7m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6b4393f582d44bb88be710426add5dd6 : SUCCESS in 13m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bddc71fc35404c5a937a941bced5d9b4 : SUCCESS in 45m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7da80d52e2464d90b7bd153b18d58e56 : FAILURE in 1h 22m 24s","accounts_in_message":[],"_revision_number":1},{"id":"fbe00f8797423b9f1088b23cda189c0645f9c1b5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-07-01 15:13:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5f3a9ac8a9c00de67eeb6171b769744b25af127a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 15:37:21.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9ec469f0503540bea42ca97aa9d5c45c : FAILURE in 21m 33s","accounts_in_message":[],"_revision_number":1},{"id":"e713ff3c39c9fb5d93b09f15e7fd5792d22c75f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 16:38:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ee8a48ab47f42cbb85e62b18a783527 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/649eb65415fa43fdbb6ce9c3c9b41b6c : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08d9cf60d656476c8113c65aa5530ab4 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa54e112c1af4cb3a4f2bc31d3e3e863 : SUCCESS in 6m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b9c4eab5ffa4de5a6be3fbbe91d6234 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/679332b5481f442dbd111a342d12adcf : SUCCESS in 4m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/839fdc5800cc4e9a8a3cfc151e5b3403 : SUCCESS in 4m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b98c80742d354a65b4c81280bb8d7241 : SUCCESS in 10m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3678473c165c4285bfc7c68acf8b600c : SUCCESS in 12m 53s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b56ca3f8fd7d42189bb371aa5ae3a296 : SUCCESS in 6m 57s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/470e6f2202864d6cbf647c95c4aafd5f : SUCCESS in 11m 30s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/838c3f808bcc453fbfc7dc6a4a491cf1 : SUCCESS in 8m 04s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d900f6fddde54b8290c3a075b9786b1c : SUCCESS in 10m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b14d6dc44a894293911491caca41e81a : SUCCESS in 8m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a6939909100e428da75dcd3e840439ac : SUCCESS in 13m 18s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8f535e77b897418db433526f8215c463 : SUCCESS in 16m 32s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5484f86bc82b4946a9c2d2cbc68e74ad : FAILURE in 9m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/51b9f1c62f3747ecab7e34d535646314 : SUCCESS in 7m 51s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ede1d851b14e42db96c0f498f58b2c5f : SUCCESS in 21m 50s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3e7211231e584797a2ad964cee65f2c1 : SUCCESS in 14m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5d5b2aa0207e446b867e0b80bd081482 : SUCCESS in 5m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2b82a5697a8d4117bcf5d343c3d36880 : SUCCESS in 24m 00s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3ce48b09befc4311b8022f966defc215 : SUCCESS in 6m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/112152e4f63345e58ed3fb4b22c19844 : SUCCESS in 8m 04s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b15cca9cc37e47eeacf9bd9c3b78bda2 : SUCCESS in 18m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2d030994b8d24e59bb40d65f719a3736 : SUCCESS in 48m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/874e01ccd5f143e0bb63be01819a1a99 : SUCCESS in 1h 23m 11s","accounts_in_message":[],"_revision_number":1},{"id":"573213fbe3d7103d853cfda2a05562c4716ac038","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-01 18:31:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b1e451f643d78e4530bee18a5e82ce9a87a443a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 18:51:49.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/450bb9f3e0554a20bb7f7e526d787ef8 : FAILURE in 18m 44s","accounts_in_message":[],"_revision_number":1},{"id":"7729b0e8bb3bb77d5e0a24a5dd25661bebf4b69d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 19:55:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3745f66b5234c6ca60bad84c350de73 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0cde7bd518743fa8b2f126790be1c65 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fdf2d923dad14ffeb1c18b287206478f : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42af32b7a78e433d80c42bf9e2b6e4ca : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ea1a78302844d89be790eda925c3440 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b0f9d68b66974454b77f3ebbf79f9a19 : SUCCESS in 6m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e84bbc51a73b4bbcb36f6d64f19e8dcf : SUCCESS in 5m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/51eeec99578546a89d1cd46653ccc37a : SUCCESS in 12m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cb28ef4f1b3845e3b8f4b864122f7ebf : SUCCESS in 12m 17s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/80897de760024c12b16590b86c98322c : SUCCESS in 8m 19s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c24f8483e9e34863902cd6b717b206cf : SUCCESS in 13m 26s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b17cca60d63e4e15a38128ddea3b7aab : SUCCESS in 7m 53s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c74891e8fd1d4ddc8780e9bb2e3ef7cd : SUCCESS in 13m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2911cbc09dce4a91a3488f14bfe28bd4 : SUCCESS in 10m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/91799a0cae074a56854857728f836ce7 : SUCCESS in 14m 22s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/849f6f6279c4491eb125bda0e9be2e15 : SUCCESS in 16m 44s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d5ada280b0184427b8a0ef4d901be255 : FAILURE in 10m 02s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/431c7c4499fb455f8fff43b5a906c30c : SUCCESS in 9m 54s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c7371458d04a4dbb90ae9293afb9554a : SUCCESS in 27m 49s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7033da8b49534dec95c578697014ba23 : SUCCESS in 13m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bf3ca267c85b425d91ec79a06470bf45 : SUCCESS in 5m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c46a579405924414b5e19025d9448114 : SUCCESS in 23m 24s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3b405750bfbc41cd83b84e3ff77ca253 : SUCCESS in 7m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/113304e254dc43ff8ec54cf9f0ed6bcf : SUCCESS in 8m 39s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7f11500ac3cc404b98bbc05d67b5bb4a : SUCCESS in 13m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed5b30947a7e489ba65dc2227174009b : SUCCESS in 48m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c652e5c88e724af4be79e4e45045521e : SUCCESS in 1h 19m 46s","accounts_in_message":[],"_revision_number":1},{"id":"e223e5aca8f06653940e059753dc965e126a8251","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-02 01:21:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b849a4a19a522464d1773ed83f972f1a14001dad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 01:39:41.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9741c56424f243e181b0d59c0e600fd6 : FAILURE in 16m 27s","accounts_in_message":[],"_revision_number":1},{"id":"eafb3d9c84cda23889f751a5d88f1577f3f16e20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 02:43:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d09d84e57b24024934a35d8af2e4129 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28fa896f5aee47e48125b182908acfb3 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c97eed00ea64d0da3a4bc68126f1245 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eaac64c87aa545369a904204205f7d15 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6059c81e33374d7f87ade27f8c7ec512 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/afe50535ca9544539a4216a11433f4ca : SUCCESS in 6m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2ea698b3f4314818b1504e9f03dea155 : SUCCESS in 5m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c816b8b054bb4d23bef3e206b4e5dd65 : SUCCESS in 11m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e06380895f5e4ac4b8caa07cce39a256 : SUCCESS in 14m 44s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/cbe9765e30a74da8916249b454bbc572 : SUCCESS in 5m 50s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6b4391cd46414405bb6b1553db80fb3a : SUCCESS in 11m 58s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2eef8ce476a947c69a67549d52eee8a4 : SUCCESS in 8m 54s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/83235befe8294124a26b683fc9eb1956 : SUCCESS in 9m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/da6ffd5033db4b46b3847a05de1e9d79 : SUCCESS in 8m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e20d4f07490a4404b3e39be555732113 : SUCCESS in 12m 12s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/74589c4948174c978f45a68a98dfef6e : SUCCESS in 18m 17s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/66175f512576403ea7c44cfbb6fe2b8a : SUCCESS in 12m 29s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8a3ebdeb6ade483e80562d54b99237d1 : SUCCESS in 8m 07s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e5c158d6b43841f3812339cf3143e93f : SUCCESS in 20m 35s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0c91b4ca97c34e78a194cdb64b3a75e2 : SUCCESS in 14m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7f9279165c3647d4b273ed9f8e204b16 : SUCCESS in 4m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1df9e4603393429ea235caf7eaae6e1c : FAILURE in 24m 25s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/87e40a303c114d3c9d1b5b6b0355d9cd : SUCCESS in 7m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9a30be7bccaa4b848f1372c7884af4d3 : SUCCESS in 7m 44s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f3baff62fc334a2fb289f1211d8215e2 : SUCCESS in 13m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9f0b127110441c3ba2466ff54ebcc3b : SUCCESS in 1h 09m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/404bd15193654f1da21e5b994e2d493c : SUCCESS in 1h 19m 29s","accounts_in_message":[],"_revision_number":1},{"id":"7962b008e9ce1aa2a4e7ae9a68a56050a98d98df","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-07-02 10:01:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"54e846b80c6c3f1f601937fdbb5d056742205d85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 10:25:36.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/483a39bad10e4514bf2db96e483ee956 : FAILURE in 20m 18s","accounts_in_message":[],"_revision_number":1},{"id":"4d75d1acda56d92617a97f2cc7ef5f684d3d8ecf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 11:13:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3035046331c244cb80cfe952fc0a0e6a : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d84f92929954b6cbba78143814b3cbe : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a95c38a84a7a46a8a6b35761677426cb : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7eee2f88cded416eb2d4ad3c61e3f4bd : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7dbc2203ce7349529dc68dabee9f7c49 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/323df54d602c4426b7953a043f580aa2 : SUCCESS in 5m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a156db6ac538454cb6c39e86c93d1309 : SUCCESS in 5m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cb17f17a04c140b781dc6375a8e8008b : SUCCESS in 11m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f72b67d36576454989bd4779dda40fc8 : SUCCESS in 11m 33s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/5ed0ef760bec4756a61037fa5b6b4346 : SUCCESS in 6m 54s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d45a7858b338464588a145df95be216a : SUCCESS in 10m 55s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3cbc3953a8f3453f81a040b63d68d6de : SUCCESS in 6m 14s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d3f494cc45aa4260975d649d85c7e015 : SUCCESS in 10m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ebb0fccd2f3449e38951c474b53e0917 : SUCCESS in 7m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b167936d2e75448e99ab34ea1636bf6e : SUCCESS in 9m 04s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/9c9194d171e44c19b23fc27d582bfa37 : SUCCESS in 13m 12s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e757d1921dea420cb61f3749d1bda95e : SUCCESS in 9m 23s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/aff871e26ea749059ada0ffa541f96cc : SUCCESS in 6m 01s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/fbaf4f3bb1f844b684b3138f033fcf0b : SUCCESS in 27m 02s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/612f8a2e6a4a4a61912d49d17443a427 : SUCCESS in 13m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/050170420a714156a62bb813e188db2a : SUCCESS in 5m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/076becdff13045a0b86368c94e491595 : FAILURE in 20m 11s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4bd0ba3464854738b25f04f430167084 : SUCCESS in 6m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d963bc1900214bf78e851854e93e2033 : SUCCESS in 6m 54s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cf5ce8270d594db69cc167c7ec656ce0 : SUCCESS in 14m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cbe7fa63715342889079cecdf7cca382 : SUCCESS in 50m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e48ee4b8b43b40d4a4181acc41dcdfdd : SUCCESS in 1h 10m 18s","accounts_in_message":[],"_revision_number":1},{"id":"6c3bf3350e43a45a50a4e64599dcbae75ddc7245","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-07-02 15:11:03.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a77db4704a637e49891a3d6ba4a6cc5d0951fcbb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 15:55:35.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e7e6aafed04645b69db3779caeea14cc : FAILURE in 22m 36s","accounts_in_message":[],"_revision_number":2},{"id":"fb05c5d6c8ea54c098e9522a2c92f4cd7f085b9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 16:44:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/036315c27a5d4609ae57efd37c98afc9 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/199c02fde032477f9685b67e172d2af7 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e1c4d04969f4e979c44236882f31b6c : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b4274775d7e4603bb974e39b3f67f01 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c03bcb9d2a449469d0a60b03092b2c4 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5b1d534147ae4d4999fad48067955df6 : SUCCESS in 4m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4fcc573c4e124d4f8c36dbd1f333e888 : SUCCESS in 5m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/43742b77ddba4a329f245a569a1749da : SUCCESS in 11m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1c5cef922c0041febaf221ad57e83ea8 : SUCCESS in 11m 01s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/0d5f27b64aec43a1bd112c962c423a57 : SUCCESS in 8m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4d70e38a739741d68afc387ce62c096d : SUCCESS in 10m 01s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/bc167c9542b94bb7a745f5c2099cc3cd : SUCCESS in 7m 07s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/918dbc5675ac4d9a818816a556eb23a7 : SUCCESS in 10m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4ed0672738cf4a3ebd4ccc14e500b819 : SUCCESS in 9m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a08d0a6bdc614536a063236462eb9d83 : SUCCESS in 9m 44s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ab97115d2fe14cf79ebaee378bc53725 : SUCCESS in 14m 52s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f609c5462fc443d7adc0821d524d1281 : SUCCESS in 10m 44s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/88f44b5ed0724e7994a2a387bb0242d7 : SUCCESS in 6m 04s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/24d09725b77f486e921601b35d43af69 : SUCCESS in 23m 14s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c120d1d0d8614595b2ce5de5dc76f2a7 : SUCCESS in 14m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c6a330506e3e4f5c903a495ba4702f97 : SUCCESS in 4m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0b0654ec991244afae45128d5358e290 : FAILURE in 18m 19s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/df44a293bd3b49cc95053f033a099550 : SUCCESS in 6m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1b3cd55e4b7c457e849a40042dcd4022 : SUCCESS in 6m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a66181281e4044419445410e75a586e7 : SUCCESS in 12m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce0d779784ab4d5ba0d6a36b9d602243 : SUCCESS in 59m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/57ffa5540a3949bca537105c1d630ca2 : SUCCESS in 1h 14m 07s","accounts_in_message":[],"_revision_number":2},{"id":"d824cac77a4e704b3d40398f956e9b56eb2da327","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-03 06:58:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"91aac4a7f0f3e577d0f92aa89d3efb7b25ad00d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 08:37:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05c1ebaf6ca0455e87ce66c61f493a42 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99713c6e54514b8980ff898a7487b297 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b835a834d7741cab34c0add19551e18 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b012877d5004ed683e1f0e46828da76 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/971f14a2ff6347738ec424fc9fce4c75 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/430f723aadd949b28f1136b4fbf809f9 : SUCCESS in 4m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a3e36c7dc1ee4f339255ca758adabeb8 : SUCCESS in 4m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/83008a77a9004f5fa7b421a584bd40d2 : SUCCESS in 10m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/87d731a368bf4d2695e545da87e433c2 : SUCCESS in 10m 25s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/45209b7949624e0388e5d29ea7385a4a : SUCCESS in 6m 14s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e4197d1d213d4e44a12ef054d5133494 : SUCCESS in 10m 57s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d179759a55f045bf9e8dbe0f881009b0 : SUCCESS in 6m 14s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c02f7158f5f64028ae7474713341aede : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/65fb1c80f53e42989dfc6006533d16c0 : SUCCESS in 8m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/809ff6ebfab1439695235f75d2b407fe : SUCCESS in 9m 52s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4fb9d2afe82f4e48a68a6882eb66ae5d : SUCCESS in 17m 10s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f4cb193ac8d04664aaa404b21026e1bc : SUCCESS in 9m 29s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/07b8f69919da421eb4090ac260f40059 : SUCCESS in 5m 37s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/de6769ab69dd417290c968d781893b4d : SUCCESS in 26m 56s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6aece0616f3540d099bdf20169ecbf71 : SUCCESS in 13m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d0513109d2cb4a01bc8a9e82a3090961 : SUCCESS in 4m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f4f74c4fb2f34f8387792befd790e842 : SUCCESS in 20m 30s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/52eb5aa7dfd7422a9acd782cd27cc978 : SUCCESS in 6m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c9875d746b714f3fbf600fde13ad912a : SUCCESS in 7m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cbce63f832844706851ce91c467ec574 : SUCCESS in 13m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b943f027d6d47bc9eebb55a757ea0d2 : SUCCESS in 47m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dd21b5d4161441c4b5c8f43a6004650a : SUCCESS in 1h 33m 40s","accounts_in_message":[],"_revision_number":2},{"id":"b2e1e732f2ad0665a1d3b24e4531899a6fe38ac4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 08:42:37.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/41fc112d17ea4fd5b6cde58c9a8533ff : FAILURE in 27m 14s","accounts_in_message":[],"_revision_number":2},{"id":"a394584304be0b146fb3967c2042018e16b5068e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-03 19:15:31.000000000","message":"Patch Set 2:\n\nthis was merged by the bot in I57e1be4fd55b9724a3e5347d3a3bd2ea3a92659e","accounts_in_message":[],"_revision_number":2},{"id":"be6b6fdb57f66b9c92635fe15180a3343420fcdc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-03 19:15:37.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"70de08cfac49f6f48c0b79a8c42c0eb42cc2e37d","revisions":{"27b83c1d1335790f1de024e3e1fdd968c835b306":{"kind":"REWORK","_number":1,"created":"2021-07-01 13:38:33.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/84/799084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/799084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/799084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/799084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/799084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/799084/1"}}},"commit":{"parents":[{"commit":"5c4901e38e5ee51cd28cf6e42ca904f19a001f47","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c4901e38e5ee51cd28cf6e42ca904f19a001f47"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-07-01 13:38:18.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-07-01 13:38:18.000000000","tz":180},"subject":"Bump osc-placement version","message":"Bump osc-placement version\n\nChange-Id: I5b30529be4f6cd732570a7b3a0446894c42387a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/27b83c1d1335790f1de024e3e1fdd968c835b306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/27b83c1d1335790f1de024e3e1fdd968c835b306"}]},"branch":"refs/heads/master"},"70de08cfac49f6f48c0b79a8c42c0eb42cc2e37d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-07-02 15:11:03.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/84/799084/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/799084/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/799084/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/799084/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/799084/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/799084/2"}}},"commit":{"parents":[{"commit":"5c4901e38e5ee51cd28cf6e42ca904f19a001f47","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c4901e38e5ee51cd28cf6e42ca904f19a001f47"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-07-01 13:38:18.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-07-02 15:10:59.000000000","tz":0},"subject":"Bump osc-placement version","message":"Bump osc-placement version\n\nNeeded-By: https://review.opendev.org/c/openstack/openstack-ansible/+/798955\nChange-Id: I5b30529be4f6cd732570a7b3a0446894c42387a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70de08cfac49f6f48c0b79a8c42c0eb42cc2e37d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70de08cfac49f6f48c0b79a8c42c0eb42cc2e37d"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
