)]}'
{"id":"openstack%2Frequirements~799088","triplet_id":"openstack%2Frequirements~master~Id74408afda36f96209ad4804b9e40d94982aa984","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"Id74408afda36f96209ad4804b9e40d94982aa984","subject":"Update constraint for decorator to 5.0.9","status":"ABANDONED","created":"2021-07-01 14:22:49.000000000","updated":"2021-08-06 08:50:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"54a064a657c838c76a222de9dade80dda3edf693","_number":799088,"virtual_id_number":799088,"owner":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-07-01 15:43:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-07-02 07:00:17.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-01 14:23:52.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-07-01 14:37:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-01 15:43:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-01 18:31:09.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"}],"messages":[{"id":"d2523b0c30c848398a3bf8f47fde5014604117d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-07-01 14:22:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce097b9ed59089ab950643b61fd884aa2eff7e25","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-07-01 14:23:52.000000000","message":"Patch Set 1:\n\nNote that, in Fedora, we had issues with decorator 5.x and some openstacksdk unit tests. Specifically, we had to skip openstack.tests.unit.cloud.test_caching.*","accounts_in_message":[],"_revision_number":1},{"id":"70188bd38c5d956648a042a08063bb028a589b2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 14:37:31.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4730042d39fe40b69877fba90f150a01 : FAILURE in 12m 37s","accounts_in_message":[],"_revision_number":1},{"id":"75ea099ace8865463ca41b43b882de85ef9ddc42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 15:43:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e01202da2654111babc92c7b0b3ec35 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/930fb681f1b546f4a18dbada77c4c75d : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/419daa22b3ae43cc8caf63bd9e4a32e3 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71e0c3dcfd9543c18a79518de208c8ac : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0fda155acd54e4180a904969ad79ae8 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/debfc0eea2754c97b30f82a6800929d8 : SUCCESS in 4m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3b12ef03dacd4d18b2ba308b88755dbd : SUCCESS in 4m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/627559813db94572bd30336e19bbeb05 : FAILURE in 6m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bd7de30342c44fc1b599535ea3d1c091 : FAILURE in 5m 58s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/244f143f91344c5ba3aef122cdcd4c60 : SUCCESS in 6m 07s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/29ec3cb9d4d641c58c6931c716204094 : FAILURE in 5m 15s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0dfa4087ad2c4b1b984453fccb6a007c : FAILURE in 5m 37s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ab3eb23f95f64cf89f201c4040163602 : SUCCESS in 10m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/87d6b09316174cfeb12b6fc7850160c1 : SUCCESS in 9m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/822a256f90004dde9690fd3d63fe6f66 : SUCCESS in 10m 50s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/aa32e96006e1400ba61e88493e0e6dc0 : SUCCESS in 13m 54s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/de913dd598324afdaa1b34181a894b6d : FAILURE in 9m 39s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/426258580b08433d91a76c5da1bfd7ad : FAILURE in 10m 21s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6ae33db5ed044b0b8dcbc0a7e3518d53 : SUCCESS in 27m 49s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ecc4669134df4ddeb09121ec9350a4cb : FAILURE in 8m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/eae9ebf04ea945f18d456b6472762fa9 : SUCCESS in 5m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/46550bbd56e740e599f6b4d855bb61a2 : FAILURE in 17m 32s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/cb585bdcdcc047fca64ac5d360377349 : TIMED_OUT in 41m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a51a67f8f4834bf49d0cb69aa294b739 : SUCCESS in 7m 57s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2946e74649b34feba13f852e40906b5a : SUCCESS in 12m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d33c4abc626a4fdea4dd1bbb5d4e6576 : SUCCESS in 45m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/09e3db0c6ea54a8a89b366e71a3cedec : SUCCESS in 1h 19m 09s","accounts_in_message":[],"_revision_number":1},{"id":"51aa4de459c32d3b37869aed33ad05ccdb998ed3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-01 18:31:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ef61eb2163d389df309f732f2dcfa35345d45de1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 18:45:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/aaf4a576d73a4e15bf506d91eb436dfd : FAILURE in 12m 08s","accounts_in_message":[],"_revision_number":1},{"id":"8dafb136c391544519d313cfa47fe42143509f52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 19:51:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da5decab05994f5b9001f4fdd74b627a : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4b13978f93640a0a772616676ac3e5c : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86d7b4c14afa433f8198045ffa651cf5 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7f25052cb91457fa23e1c45cdedbf63 : SUCCESS in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f353923c5e6497587152b6a0caa2b66 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7aeb529a83b741b6b846e0f6d2ddc406 : SUCCESS in 5m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bc92540b790d4d95b0402f36a8477243 : SUCCESS in 6m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/10ae1a37805243369640570d32c29ce1 : FAILURE in 6m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/02e391f8bcb0475387b2ba8506c85341 : FAILURE in 7m 27s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e17d68bc8f004dfd851577f953bd7847 : SUCCESS in 9m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f383f31ff8e24966913e5b96081d0a36 : FAILURE in 7m 36s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/71e300829a4341a398d6523660afa5fe : FAILURE in 6m 11s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/00daf306a55a4ebda95adb54b2deccff : SUCCESS in 12m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4fbdb16aa12e4474af3e9d5b46e0a788 : SUCCESS in 11m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d3aab1efa5294727a5e67e11db3ca730 : SUCCESS in 12m 26s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/08a0938a9b7a4e86bc2e85b12e173f42 : SUCCESS in 14m 59s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6b9de891957c42abb9b5923186c531e1 : FAILURE in 10m 26s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/574e9e2b94d0435797a6491ac48c36a4 : FAILURE in 10m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/bf317374e1c5413290a55ee842e98b52 : SUCCESS in 32m 02s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c8e42ce353674d5bbd4165a54918c5ff : FAILURE in 11m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/53ed808b46fb4ded92b4a9269cb39ecf : SUCCESS in 6m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8aa58e63031b4f188c9b06f1d7a26f0a : FAILURE in 27m 49s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/bd6563480ce94d43a44435bc8cc1d0f7 : TIMED_OUT in 41m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/69f9751dc9454083afb42f6c855dcb66 : SUCCESS in 8m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/079acbd8accf48558db8f629d9185f61 : SUCCESS in 15m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f4d088bbe982431da01e6e2ea0f92a67 : SUCCESS in 49m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da6d86371f7c4ba0bf985326e480ef78 : SUCCESS in 1h 17m 13s","accounts_in_message":[],"_revision_number":1},{"id":"0d026d25ea406a4e5128bdc03961f3c1c5bfa129","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-02 01:21:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0285330e8cb882bcb2bd6ea2c675fd48721ccf4d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 01:33:38.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a15c43c20e6e45749e08acc872acdd73 : FAILURE in 10m 34s","accounts_in_message":[],"_revision_number":1},{"id":"8f65aa1d4b6c20c8a6990048be2651b660044c82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 02:42:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81f12ca4ea234eb3af9897c3825b0cd3 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/257d7dcba15742ae8d0e5d7e7985bc2b : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e2e40cf864c4801b3911e602269e948 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d65c39b4fa1481f8c45dd5f4c00e345 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b48fd7212334f179e1034b358af9bd1 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/016a6f1d63684a4993c144a0cdee1106 : SUCCESS in 4m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/40f1de0c4f864305b4f8c27938bc61ec : SUCCESS in 6m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ba011486372c4dca80e4957e84ccb2e1 : FAILURE in 13m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6a2be095fce44e649782e4b847ba5623 : FAILURE in 7m 43s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/28507a2a724341a4a0a9991273b66936 : SUCCESS in 7m 42s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/66ae645671074c938755a2fbe2e83560 : FAILURE in 5m 31s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5b661d9b20f949238913bb6cda4a4573 : FAILURE in 7m 29s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d342240a9b17429c8f435c7856df94be : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c14d55a27e4d49f2816ff568c96fd0ef : SUCCESS in 15m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e4ec7ef3fe624f7aa388003f081afd38 : SUCCESS in 10m 02s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d869044264f94ad9bf45170b70a7dd32 : SUCCESS in 28m 37s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/13f273e26d0b4aa8bf21dd4f3e348a9a : SUCCESS in 13m 14s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/17a0759d16d544e3849182f3df18bad3 : FAILURE in 13m 16s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9bcf70b922a44ad59619417e03084422 : SUCCESS in 28m 05s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/29f8022bd1c04aed8159b795264d6115 : FAILURE in 9m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/81d4c339b75a49908556c56ecb978852 : SUCCESS in 3m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f1f46eed2ee547e4b3bea625c974f74e : FAILURE in 18m 26s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2da8f4eff53d4e0f8247d1214b0231e9 : TIMED_OUT in 40m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dd492babd81e4e598ebe2516815357ea : SUCCESS in 7m 30s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7dbc9eac952445089318e6349bbf6946 : SUCCESS in 11m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eab5dc5973a742d997220070f30e6f73 : SUCCESS in 51m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/15b36b200f6e4dc5ab30080930d439ce : SUCCESS in 1h 19m 39s","accounts_in_message":[],"_revision_number":1},{"id":"c405b953994cf778e2eeeebe492fadb645ffa821","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-07-02 07:00:17.000000000","message":"Patch Set 1: Workflow-1\n\nSo the check-uc jobs fail because of the following:\n\nThe conflict is caused by:\n    The user requested decorator\u003d\u003d\u003d5.0.9\n    networkx 2.5.1 depends on decorator\u003c5 and \u003e\u003d4.3\n\nI see networkx has recently removed decorator from the requirements [1], I think we\u0027ll have to wait until we have a new release.\n\n[1] - https://github.com/networkx/networkx/commit/89a54703fafbb67d5a79b38238f7478ef62a51eb","accounts_in_message":[],"_revision_number":1},{"id":"e0153c8e848b653e203d46d01dc3ef89f1afe525","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-03 06:58:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"629b5386f4037537212018bd1c66475e628f5ff3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 08:16:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90cad86d3dfd4960a319eb91b99e1307 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/373d29382840434bbcb9c60486297bb1 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd999ce53e534df88c72ece753fda116 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5844b56def654d9bbfea4f078ef7ec55 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2c201e5e8e746b7ac4451312752824e : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b50cdc78cbb94b3098f5c807e5f8fd10 : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a3fdf32f2eec4d7a99c425517f40c6d5 : SUCCESS in 4m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/843c7ea74ed945fb92e5e2e664f6252f : FAILURE in 5m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/09fe7379b42a4c2e84ba95566fd9966c : FAILURE in 5m 15s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/2cae94889fb7442788cd9b4a813de4f7 : SUCCESS in 5m 46s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d929a10b3cf344c2811c82edeb053ebc : FAILURE in 5m 07s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/cba86d9121f7490bb83ab7b13d25ff59 : FAILURE in 4m 29s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/203994cc7da34496abf671f1fd2aff9c : SUCCESS in 10m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/43f42b2422734edaa1139ed460feee22 : SUCCESS in 9m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/21155e9402a446c98457b305ab33b24b : SUCCESS in 8m 21s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/42b5aee72a834213a42310c38e05581a : SUCCESS in 15m 35s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/56ebecaecb7c45e6afbec266d458e29a : SUCCESS in 8m 53s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9e2ea02ac5c74d7c879c661ebdc4ac06 : FAILURE in 10m 57s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/bbb75d3df9a24353a13a4aedae1ae536 : SUCCESS in 23m 47s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/63ce60cac52f4859aafe2ee53d5677b6 : FAILURE in 8m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d6e58bba7412436aa333b12d797d396e : SUCCESS in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4279bf6243ff4da3bea2388fc3db27d2 : FAILURE in 26m 55s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4e170525d6a347fb90b9b16ad70258ac : TIMED_OUT in 41m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e079fbfb117a444ca2153c7f06019204 : SUCCESS in 6m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/27ee9010a3a94856830c8dbac85c1cf6 : SUCCESS in 12m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7db6ae04b9004a2d87d2402121e0246e : SUCCESS in 53m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e8708f1166384af2a3b29a228d4e9116 : SUCCESS in 1h 14m 01s","accounts_in_message":[],"_revision_number":1},{"id":"e7ffe344939c5cd9d6abf77445dbb8c3605f1f36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 08:34:42.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0e6f36826bd1471984058df81685cc3d : FAILURE in 14m 52s","accounts_in_message":[],"_revision_number":1},{"id":"0d63077e2b6566e2a45be4d1248dee5002ff8e60","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-03 19:16:04.000000000","message":"Patch Set 1:\n\nrecheck\n\nguessing it\u0027s legit...","accounts_in_message":[],"_revision_number":1},{"id":"7d49c8c09ebfe84501d910437550a82160d66a1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 19:26:55.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c0fae82e808b4c849ca5e1e54dbb5905 : FAILURE in 9m 18s","accounts_in_message":[],"_revision_number":1},{"id":"2aa0bb7b7af2dcb5727a38e22476095a4bc08e4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 20:32:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14abc4f409984273b3311650b5e26cf9 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5e79ed2901248e3b3e4782fd6755801 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d9ee4be6b224dbfb98ec429cb88b4ab : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95b912cce16248b5888fab75fe9c6c74 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43d67c55156f415aa1e4080531e787e4 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/87b09701f1de4d2cb9ea7e425a91beb7 : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/76d8450b8e9e4b339cff24927a255363 : SUCCESS in 4m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/65885ab16fde46cda7b90d4984a84531 : FAILURE in 5m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1515c697e2a74d30b96bbc580ea48317 : FAILURE in 9m 06s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b52712dac43e4ce0bc1eed6f4882358e : SUCCESS in 5m 18s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ab235cd370414cb29faf47404ba168d8 : FAILURE in 6m 05s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/18e909e3b944438ea3f994338429ffd9 : FAILURE in 4m 01s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9df7de6c161144ff9b839c73f2d27105 : SUCCESS in 11m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/99547276a05d4f55b076d73cbfea44de : SUCCESS in 11m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4b4d38b0db9548b7a73d90c18ba4fc28 : SUCCESS in 8m 52s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/fb4229f1d9f14f4c8393b09eb5ef2f73 : SUCCESS in 18m 58s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5ff88f0cfc434da19f6e43811bb9ee3f : SUCCESS in 10m 30s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b3e998a3f490471788f727c8c2bd5708 : FAILURE in 7m 24s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/bef9ae8d4a38485f8cd83d2bb0d19473 : SUCCESS in 21m 27s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2304a943babe4e4abb1621791e68c1d9 : FAILURE in 7m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ebd4bbdabc6d4f64875dce1e5b2dea06 : SUCCESS in 5m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dee124099c1c4e93839712baca02156d : FAILURE in 16m 39s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/449778ff1ab042a68651fc4a47a5605e : TIMED_OUT in 41m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3adeffaf884142b689f94401b8d090ea : SUCCESS in 6m 00s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/568684ac88764f80aee108c173a0501f : SUCCESS in 12m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb6f1ae733bb4eda98e43c596d91af55 : SUCCESS in 47m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f93aafaf07684901b21edb8f3e64437a : SUCCESS in 1h 14m 59s","accounts_in_message":[],"_revision_number":1},{"id":"54a064a657c838c76a222de9dade80dda3edf693","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-08-06 08:50:41.000000000","message":"Abandoned\n\nAlready covered by https://review.opendev.org/c/openstack/requirements/+/803113","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"77ae7135b7fd82e9c8b9a6544e16f424cda3ef15","revisions":{"77ae7135b7fd82e9c8b9a6544e16f424cda3ef15":{"kind":"REWORK","_number":1,"created":"2021-07-01 14:22:49.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/88/799088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/88/799088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/799088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/799088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/799088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/88/799088/1"}}},"commit":{"parents":[{"commit":"5c4901e38e5ee51cd28cf6e42ca904f19a001f47","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c4901e38e5ee51cd28cf6e42ca904f19a001f47"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2021-07-01 14:21:30.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2021-07-01 14:22:47.000000000","tz":120},"subject":"Update constraint for decorator to 5.0.9","message":"Update constraint for decorator to 5.0.9\n\nThis was released in May, but we never saw a bump proposal.\n\nChange-Id: Id74408afda36f96209ad4804b9e40d94982aa984\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/77ae7135b7fd82e9c8b9a6544e16f424cda3ef15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/77ae7135b7fd82e9c8b9a6544e16f424cda3ef15"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
