)]}'
{"id":"openstack%2Frequirements~799315","triplet_id":"openstack%2Frequirements~master~I608331e28ef349399068f9b1fa16bd6a6c5779e6","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I608331e28ef349399068f9b1fa16bd6a6c5779e6","subject":"Add datetimerange library to global requirements","status":"MERGED","created":"2021-07-02 18:01:58.000000000","updated":"2021-07-26 05:03:55.000000000","submitted":"2021-07-26 05:02:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799315","meta_rev_id":"32add32c21cb3675ec0f1bb778676aee364ba6ac","_number":799315,"virtual_id_number":799315,"owner":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-26 05:02:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-07-26 03:35:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-07-26 03:35:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-02 18:33:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-02 19:26:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-03 06:58:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-07-03 19:18:08.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"487e71381a523fe728053dac83d62ccafe11445c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2021-07-02 18:01:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ce0e1bac4ec77b397cffe5ffe4be922cb8876c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2021-07-02 18:09:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"58890c5fa048fda02e5e6877b506a8fd7875f77b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 18:33:19.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/784f74f2863047439612f1c6d50df0f7 : FAILURE in 21m 32s","accounts_in_message":[],"_revision_number":2},{"id":"30a3f4d1997505ed2bb70ab7d36902390188ff4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 19:26:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b10cea2876f040a38a8a358a082f89f9 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9681fe399ce4c5d8d86214d82cabfd4 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0f5014e6bf74e82a044c2f555f4cab1 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45079c0778a04b3dab1e591bc0058a31 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ddbe0dd285214005bfd0aba94ff46bf9 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cb7905164d7249cf8274b2e912c9a82c : FAILURE in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/288b96aeba8249888421f3002ff9ebeb : SUCCESS in 4m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/aaeca56a95aa4a4189379693ebc924c1 : SUCCESS in 11m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/49149ef060ae4ece922e744179f7ecc9 : SUCCESS in 16m 07s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/7d82a0923c7643638036502c45047571 : SUCCESS in 5m 48s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/85d588a4f1854e6aa96963f496c5d89a : SUCCESS in 10m 14s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f44d9fffe97149a8be8b8268ad0f6911 : SUCCESS in 6m 15s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5f01328ad5e14bf29101d6332f17f50b : SUCCESS in 9m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c9f3a96a67a04e96b1c98728e51350e9 : SUCCESS in 7m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c1639a68fb0a480aacd331f31c231878 : SUCCESS in 9m 45s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/bcce4f8106644f85b895057a2a65bc95 : SUCCESS in 18m 37s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8765f353efe84853affdb174e973ef7e : SUCCESS in 17m 17s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a9290d65fa7d402f831d77f6b4b35097 : SUCCESS in 5m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e63cf2fa4323487db71be78182c329e1 : SUCCESS in 21m 46s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/de28875924a142bc8c52906032223462 : SUCCESS in 12m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/325290a1ac3243c4958d10c185cdb353 : SUCCESS in 4m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/698839d4aa0f4deda9d7a1b39a0b1656 : FAILURE in 21m 47s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/aea409f1f77d4019a85f45a15904f61b : SUCCESS in 5m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ddbf0588b44a4e5c89c044fa39f1833d : SUCCESS in 6m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/de90a3d6541d4d449df75335e70bf23e : SUCCESS in 19m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30698e90fcbc4afdba8f4ff219b65c67 : SUCCESS in 44m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f9a5c11e4894b81a097632eca4fab17 : SUCCESS in 1h 15m 54s","accounts_in_message":[],"_revision_number":2},{"id":"ee08a434ad91bf40b4b2e6f48bccc9856045326b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2021-07-02 20:02:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"76e653d06d292e500d4ac8294c6728ffbcc1205b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 20:43:21.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/57a06c3005d545d28dfbc0d09a274118 : FAILURE in 23m 09s","accounts_in_message":[],"_revision_number":3},{"id":"ab75c0f52ba8c1639c97d923cf1d6a653469c9bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 21:10:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0d61835f77a46a385cce45ba50e26b6 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad558c7416754b818cd9432686a8bf3a : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/598320e5de894f59a9408de7498ee6af : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28ca3610567c496aa7dd783cd57b9f9a : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/554a45851cc64b6abd777ca063eb341f : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9308887645b849b1b63dccec49f0bf41 : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/baea945d3c7a46bb8006439672c92d4d : SUCCESS in 3m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0cd89ac26c5f42ec86ff00538bae6f07 : SUCCESS in 10m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d9600a623bbc45ad85c855a58ebfebeb : SUCCESS in 11m 40s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f2a67e64dcfa407ab49bae43ac11f413 : SUCCESS in 6m 11s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a20d195351c94199ba243fa2ce124243 : SUCCESS in 12m 29s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/80ce6c0c91f34ce98906968ac2fd4783 : SUCCESS in 6m 37s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/13c25c4cf1e34091ae2e6823b71fc9e6 : SUCCESS in 10m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3f9ab3381a614327a0008c8d25be7bf6 : SUCCESS in 8m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6857363ac9ce4824b1ee950d36b7b319 : SUCCESS in 12m 56s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/595e6d27bfe4405cb3149cdbf7f80914 : SUCCESS in 14m 37s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/764dee2931d542cd8a8d8979d1981330 : SUCCESS in 8m 48s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e858d4fa6e7045619aa52631ebadb063 : SUCCESS in 6m 05s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7bfb3e00b2ea4c65b81363e66bfa395f : SUCCESS in 28m 54s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/952ea094c07e4104bac60e8a997af986 : SUCCESS in 12m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a58c66e48097451884b15baf2bcd61d9 : SUCCESS in 4m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0366ed67880b47be8857941dbd77f8fc : FAILURE in 18m 04s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/73c732fd59c740818ef0d26113479e0f : SUCCESS in 5m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8d7acdd10dec438ba99e6f664b3469ad : SUCCESS in 7m 19s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e11dfca4e0974722b6ac2f0866c42be4 : SUCCESS in 12m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af8702e2dc8946de90a243a552774a8c : SUCCESS in 48m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/af96bca0aa7c472fb7d14e2f87719ea0 : SUCCESS in 1h 07m 21s","accounts_in_message":[],"_revision_number":3},{"id":"b9452b39376f1d9b95f33b419d6420464bad9059","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-03 06:58:14.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"42af5c7deade54a8aa3b20352edbe053bbbc20cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 08:36:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a039d7006145497caed1df25c5391e34 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d06927a52e7c4ec2b8de9006d3812385 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c252561a4fbe44db87dbd22288fafd92 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2dbe3052394d4b65bb3006d59beaea99 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29f3cf9c279c41f895298af44da25270 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b44b1e823c0443ef804622236b01508e : SUCCESS in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e26988f4b5144e7aad2d8040c7f39cbc : SUCCESS in 4m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/01d4e97829c642f2a4464c4213fd5f3d : SUCCESS in 10m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0266667691e642d7bc404c9ffcc3d7d9 : SUCCESS in 10m 12s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/181601a72b22445ba4244220a6a1e2d7 : SUCCESS in 6m 31s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d3b2ac7a68e94b27b993866219efa73c : SUCCESS in 10m 06s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/fe53830de89d490e83a039459c3a721e : SUCCESS in 6m 08s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8751214561544177bb1d3e0188bc3e3a : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a3c77f4535304f079154d244ac65c48c : SUCCESS in 8m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f420fcd8d27440b28af875af9e7bc64c : SUCCESS in 8m 43s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/9bde000497764c9dafc8264d97cd85d9 : SUCCESS in 15m 52s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/644ebe1965774384baebe03f72e7832e : SUCCESS in 9m 23s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/3da986f75e2f484bb0fdd738c5ff75c4 : SUCCESS in 5m 27s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9923771335b4405f84c6631e9ac2b40f : SUCCESS in 24m 06s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5c31e6efab6440478272d503426abef7 : SUCCESS in 13m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/256f077ce2274dfa9a89b2eca28a30e7 : SUCCESS in 4m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e550a906f23e47a98ed9a1e67f67646c : SUCCESS in 21m 14s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/88cbc1734393460d91cfb78ef640d0f3 : SUCCESS in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/47783b9b0ffc4083ba38877237534ae7 : SUCCESS in 7m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f5c6b27ca6a54738830b80e898ac3703 : SUCCESS in 14m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54f396e0aad442c98d045e7c7caeb634 : SUCCESS in 52m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa1e7ffca5534289a49d53e33d8f2bb3 : SUCCESS in 1h 32m 35s","accounts_in_message":[],"_revision_number":3},{"id":"3cd7282c416b4387e283d29a74d1c06ce9409366","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 08:39:31.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d2e4f0b417ba44009cc3633f6c4385f6 : FAILURE in 19m 31s","accounts_in_message":[],"_revision_number":3},{"id":"6770a3862ea87f23f12cd1613f5f117c3ed417cf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-03 19:18:08.000000000","message":"Patch Set 3: Code-Review-1\n\ncan you please update the commit message with answers to the linked questions?\n\nhttps://docs.openstack.org/project-team-guide/dependency-management.html#for-new-requirements","accounts_in_message":[],"_revision_number":3},{"id":"3173f8d14c8b40fbf6baaa8cdad830b6cd510e3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2021-07-05 13:29:58.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"c357504e174b223c49faf11b69ab635d98a332cc","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2021-07-05 13:30:03.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3: Code-Review-1\n\u003e \n\u003e can you please update the commit message with answers to the linked questions?\n\u003e \n\u003e https://docs.openstack.org/project-team-guide/dependency-management.html#for-new-requirements\n\nDone. Thanks for the review.","accounts_in_message":[],"_revision_number":3},{"id":"30f15a3bfb50c83620556b57ff9d722dda97118b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2021-07-05 13:30:54.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"5fb30d57be4603de5db2f2eb17e88020fa28ef3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 13:50:48.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f86e4233fd564c52aafbe8c16d17fbd2 : FAILURE in 18m 08s","accounts_in_message":[],"_revision_number":5},{"id":"9f98eea6ea9326b8867c615677b0e82b95402712","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 14:47:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77f2e26b91b24aac9b690e7bb2c441c8 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/606a5a21a7a64bfb89916386603959bd : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba24888e64a2447090ed23d6ff2f7142 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d5ca32f69ea4b298cffe44ae877fae9 : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22fc4331e9f14bd5ab6a13a32cab2621 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/62f98f2fac4743b4a3f7b957f1b22f9b : SUCCESS in 4m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e9a14769bb024e6ca4448dc5cac75108 : SUCCESS in 4m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c04e4eadf74143e1ac0c1350d15ded8f : SUCCESS in 12m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c49fc4f8b9384f489ec476ac63f23624 : SUCCESS in 10m 09s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/0820be50d42b4cef9588500fe35c6133 : SUCCESS in 6m 28s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/42e993ed1dd74a2ea84a664fd72bee76 : SUCCESS in 10m 24s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5c7cba0b18e443948c36041ded6ff684 : SUCCESS in 6m 35s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3ecf096159514c8aa6d8d147ae2a45f3 : SUCCESS in 9m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/348b9892c8934c9c91480a1ffb97fe3b : SUCCESS in 9m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b3b5c19cb59041b08a58b7cdde9de35a : SUCCESS in 10m 04s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0b9d4d5b646f4c139bfdc62bd5b7283f : SUCCESS in 16m 47s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c159ff320c814e6e869c015761d6a932 : SUCCESS in 12m 47s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6f87efc4a1f7464c83731e8af700123c : SUCCESS in 6m 37s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c6f09b04aaf944978897a51cc1d735f5 : SUCCESS in 21m 19s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2aabd1eca4b54dbdac38ba7f8cd9f10c : SUCCESS in 13m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e9ee0336647a4e59a0f8a9fbdb57b9e9 : SUCCESS in 4m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/413c075f961543df8d0f40281986c76e : SUCCESS in 18m 27s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/755880ee19b644569bebfd5f081b9956 : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2afccedded564fdb8fbd6488f4526282 : SUCCESS in 7m 15s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7585f88d9b184762aa571d6355f72991 : SUCCESS in 14m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/984d894a92fd4160a6af4b3c2073008e : SUCCESS in 42m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1271921307f9443abdd90772ae2cd888 : SUCCESS in 1h 15m 35s","accounts_in_message":[],"_revision_number":5},{"id":"2eb7e8d865ebdc6c0e9b008a8bfcd5d884762140","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-06 04:37:49.000000000","message":"Patch Set 5:\n\nIs there a reason this needs to be a library, there seems to be a bunch of things you can do natively in python for datetime ranges?  https://stackoverflow.com/questions/993358/creating-a-range-of-dates-in-python  I will say the library does seem cleaner, just don\u0027t want to turn this into nodejs :P","accounts_in_message":[],"_revision_number":5},{"id":"e3a25f9cbe2e087909d994154bc000c436b7d62a","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2021-07-06 11:11:07.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 5:\n\u003e \n\u003e Is there a reason this needs to be a library, there seems to be a bunch of things you can do natively in python for datetime ranges?  https://stackoverflow.com/questions/993358/creating-a-range-of-dates-in-python  I will say the library does seem cleaner, just don\u0027t want to turn this into nodejs :P\n\nI read the same stackoverflow and other places. Also the same question that a friend posed. Well, if we write the code, it is more to maintain, document, and to guarantee that it works (via unit testings normally). By adopting a library, I know that there is a solid solution, tested, and used out there. Moreover, it is less code to maintain throughout time.\n\nI personally see no reason to replicate others work. So, if there is already a code (library) that does exactly what I need, it is well maintained, the code is neat, documented, I do not see the rationality not to adopt it.\n\nIf people are opposed to adopting it, I would rewrite it, or part of the solution in CloudKitty. However, I do not see that as the best approach, specially in the opensource world.","accounts_in_message":[],"_revision_number":5},{"id":"2e57a8df528a3f84eb2f11c9163ccee13884faf2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-26 03:35:34.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"07de5938d7065eef9a48c59dc65b9b0b3895ccda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 03:36:09.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ae571552515faff6bde49202ceec9c6546c88912","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 05:02:42.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cadf338ff8f4ea79f0511ff17bff0da : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd44607bd38d4feabbb94eee86db3c8e : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42c264a1b7ef40b0a2524853b507c0f9 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e9bd4ddf41c4d2b99f364be8853777d : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f3672fbb604e45bd934cb4065622d285 : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9e0a941905554f0e862e9171aca54d22 : SUCCESS in 4m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f2d64b92c57244109fb6177140476eab : SUCCESS in 10m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7e6e9dd16ff44a01b27ead0587351b59 : SUCCESS in 9m 09s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f3c1c8d9792940ae8b732ebf3fd1bd9e : SUCCESS in 8m 40s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/188656f856ff433d929c90de82e3ecf3 : SUCCESS in 6m 06s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/53a7af3232594090a041cf28560542fe : SUCCESS in 11m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fb6d346c2a1349b9929a28fd1eaad5cd : SUCCESS in 8m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/aa472b4a2df045558fa1900eb465acd4 : SUCCESS in 8m 29s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/92a20e0e67544b29ac920c8aad655a05 : SUCCESS in 13m 56s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5578505faaf54cab9d21ac96696fd19c : SUCCESS in 8m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/58323906628e4fa892368ae3e0c37782 : SUCCESS in 5m 22s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/709d670b8f2e45c79ee93e5d8a822525 : SUCCESS in 27m 16s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7516512fad8148a4b7901ec8cf80ecca : SUCCESS in 14m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a7541e056491451ab1438fe884ac3e53 : SUCCESS in 3m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/de22bde14b744e2482221775fc90e754 : SUCCESS in 18m 07s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6729111f2f3e475590fd7398e2bae82b : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0e2c232960a9499fb3ba9c68914ced4c : SUCCESS in 6m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ed4a11186d9c4665b8a126b1cdc8136f : SUCCESS in 14m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9532d4ae9124d5b99f0388cd81fdb8d : SUCCESS in 43m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f0a6cdd880d640559adf8627366a8686 : SUCCESS in 1h 21m 15s","accounts_in_message":[],"_revision_number":5},{"id":"561d27c347d28e9c771264623ccee236bd85798f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 05:02:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"32add32c21cb3675ec0f1bb778676aee364ba6ac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 05:03:55.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88881aa7d7c54e7083ee62d411b8d5f6 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"71a8c3bd4f8d665d569fa94b113be475ccfb9879","revisions":{"62cbce148ee598f9e385965110e3ee0f4be85f6a":{"kind":"REWORK","_number":1,"created":"2021-07-02 18:01:58.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/15/799315/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/799315/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/799315/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/799315/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/799315/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/799315/1"}}},"commit":{"parents":[{"commit":"d3eb093041ca83eed0f8eef68299e2c9b07c3453","subject":"Merge \"update werkzeug and Flask to 2.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d3eb093041ca83eed0f8eef68299e2c9b07c3453"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2021-07-02 18:00:54.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2021-07-02 18:01:55.000000000","tz":-180},"subject":"Add datetimerange library to global requirements","message":"Add datetimerange library to global requirements\n\nWhile developing a new feature in CloudKitty, we adopt a library\n(datetimerange) to facilitate the handling/processing of\ndatetime ranges. Therefore, to make the Zuul requirements check\nhappy, we are introducing this new dependency into the global\nrequirements file.\n\nChange-Id: I608331e28ef349399068f9b1fa16bd6a6c5779e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/62cbce148ee598f9e385965110e3ee0f4be85f6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/62cbce148ee598f9e385965110e3ee0f4be85f6a"}]},"branch":"refs/heads/master"},"322a148bffa4a92066c7bca231403df1ac5c0d01":{"kind":"REWORK","_number":2,"created":"2021-07-02 18:09:14.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/15/799315/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/799315/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/799315/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/799315/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/799315/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/799315/2"}}},"commit":{"parents":[{"commit":"d3eb093041ca83eed0f8eef68299e2c9b07c3453","subject":"Merge \"update werkzeug and Flask to 2.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d3eb093041ca83eed0f8eef68299e2c9b07c3453"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2021-07-02 18:00:54.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2021-07-02 18:08:58.000000000","tz":-180},"subject":"Add datetimerange library to global requirements","message":"Add datetimerange library to global requirements\n\nWhile developing a new feature in CloudKitty, we adopt a library\n(datetimerange) to facilitate the handling/processing of\ndatetime ranges. Therefore, to make the Zuul requirements check\nhappy, we are introducing this new dependency into the global\nrequirements file.\n\nChange-Id: I608331e28ef349399068f9b1fa16bd6a6c5779e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/322a148bffa4a92066c7bca231403df1ac5c0d01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/322a148bffa4a92066c7bca231403df1ac5c0d01"}]},"branch":"refs/heads/master"},"a810bfbe04bf6a921cbfcf206eb25602272a4fe1":{"kind":"REWORK","_number":3,"created":"2021-07-02 20:02:06.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/15/799315/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/799315/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/799315/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/799315/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/799315/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/799315/3"}}},"commit":{"parents":[{"commit":"d3eb093041ca83eed0f8eef68299e2c9b07c3453","subject":"Merge \"update werkzeug and Flask to 2.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d3eb093041ca83eed0f8eef68299e2c9b07c3453"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2021-07-02 18:00:54.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2021-07-02 20:01:54.000000000","tz":-180},"subject":"Add datetimerange library to global requirements","message":"Add datetimerange library to global requirements\n\nWhile developing a new feature in CloudKitty, we adopt a library\n(datetimerange) to facilitate the handling/processing of\ndatetime ranges. Therefore, to make the Zuul requirements check\nhappy, we are introducing this new dependency into the global\nrequirements file.\n\nChange-Id: I608331e28ef349399068f9b1fa16bd6a6c5779e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a810bfbe04bf6a921cbfcf206eb25602272a4fe1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a810bfbe04bf6a921cbfcf206eb25602272a4fe1"}]},"branch":"refs/heads/master"},"7eeeb8809d2522accaac9eec095106b16cbd575d":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-07-05 13:29:58.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/15/799315/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/799315/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/799315/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/799315/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/799315/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/799315/4"}}},"commit":{"parents":[{"commit":"d3eb093041ca83eed0f8eef68299e2c9b07c3453","subject":"Merge \"update werkzeug and Flask to 2.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d3eb093041ca83eed0f8eef68299e2c9b07c3453"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2021-07-02 18:00:54.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2021-07-05 11:31:56.000000000","tz":-180},"subject":"Add datetimerange library to global requirements","message":"Add datetimerange library to global requirements\n\nWhile developing a new feature in CloudKitty, we adopt a library\n(datetimerange) to facilitate the handling/processing of\ndatetime ranges. Therefore, to make the Zuul requirements check\nhappy, we are introducing this new dependency into the global\nrequirements file.\n\nAdd datetimerange library to global requirements\n\nWhile developing a new feature in CloudKitty, we adopt a library\n(datetimerange) to facilitate the handling/processing of\ndatetime ranges. Therefore, to make the Zuul requirements check\nhappy, we are introducing this new dependency into the global\nrequirements file.\n\nHere follows the answers to the new library requirement questions:\n\n1 - Is the library actively maintained?\n\nYes, it is. As we see in [1], the component has more or less one\nyearly release. It is a very focused library, so this release process\nmakes sense.\n\n2 - Is the library good code?\n\nYes, it is. As we can see in [2], it has a 95% of code coverage.\nMoreover, while navigating in the code repository, we can see that\nmethods are well documented, and coded.\n\n3 - Is the library python 3 compatible?\n\nYes, it is. That is also found in the README of the project in [2].\n\n4 - Is the library license compatible?\n\nAccording to reference [3]:\n\n```\nIn order to be acceptable as dependencies of OpenStack projects,\nexternal libraries (produced and published by 3rd-party developers)\nmust be licensed under an OSI-approved license that does not restrict\ndistribution of the consuming project. The list of acceptable licenses\nincludes ASLv2, BSD (both forms), MIT, PSF, LGPL, ISC, and MPL.\nLicenses considered incompatible with this requirement include GPLv2,\nGPLv3, and AGPL.\n```\n\nTherefore, the project fits nicely into these requirements. The project\nin question is licensed under MIT, as we can see in [2], in the right\ncorner part of the page, and also in the LICENSE file at the root of\nthe project.\n\n5 - Is the library already packaged in the distros we target (Ubuntu\nlatest / Fedora latest)?\n\nIt is installed via PIP; PIP is available in major distros. Therefore,\nit is possible to use this library in all of them by issues the command\n`pip install DateTimeRange`.\n\n6 - Is the function of this library already covered by other libraries\nin global-requirements.txt?\n\nNo. At least, I did not find it.\n\n7 - Is the library required for OpenStack project or related dev or\ninfrastructure setup? (Answer to this should be Yes, of course) Which?\n\nYes, it is. We have a new feature being developed ion Cloudkitty in\n[4], which needs this library to handle DateTime ranges.\n\n[1] https://github.com/thombashi/DateTimeRange/releases\n[2] https://github.com/thombashi/DateTimeRange\n[3] https://governance.openstack.org/tc/reference/licensing.html\n[4] https://review.opendev.org/c/openstack/cloudkitty/+/799207\n\n8 - If the library release is managed by the Openstack release process\ndoes it use the cycle-with-intermediary release type?\n\nIt is not managed by the OpenStack release process.\n\n9 - Do I need to update anything else?\n\nNo.\n\nChange-Id: I608331e28ef349399068f9b1fa16bd6a6c5779e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7eeeb8809d2522accaac9eec095106b16cbd575d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7eeeb8809d2522accaac9eec095106b16cbd575d"}]},"branch":"refs/heads/master"},"71a8c3bd4f8d665d569fa94b113be475ccfb9879":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-07-05 13:30:54.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/15/799315/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/799315/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/799315/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/799315/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/799315/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/799315/5"}}},"commit":{"parents":[{"commit":"d3eb093041ca83eed0f8eef68299e2c9b07c3453","subject":"Merge \"update werkzeug and Flask to 2.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d3eb093041ca83eed0f8eef68299e2c9b07c3453"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2021-07-02 18:00:54.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2021-07-05 13:30:21.000000000","tz":-180},"subject":"Add datetimerange library to global requirements","message":"Add datetimerange library to global requirements\n\nWhile developing a new feature in CloudKitty, we adopt a library\n(datetimerange) to facilitate the handling/processing of\ndatetime ranges. Therefore, to make the Zuul requirements check\nhappy, we are introducing this new dependency into the global\nrequirements file.\n\nAdd datetimerange library to global requirements\n\nWhile developing a new feature in CloudKitty, we adopt a library\n(datetimerange) to facilitate the handling/processing of\ndatetime ranges. Therefore, to make the Zuul requirements check\nhappy, we are introducing this new dependency into the global\nrequirements file.\n\nHere follows the answers to the new library requirement questions:\n\n1 - Is the library actively maintained?\n\nYes, it is. As we see in [1], the component has more or less one\nyearly release. It is a very focused library, so this release process\nmakes sense.\n\n2 - Is the library good code?\n\nYes, it is. As we can see in [2], it has a 95% of code coverage.\nMoreover, while navigating in the code repository, we can see that\nmethods are well documented, and coded.\n\n3 - Is the library python 3 compatible?\n\nYes, it is. That is also found in the README of the project in [2].\n\n4 - Is the library license compatible?\n\nAccording to reference [3]:\n\n```\nIn order to be acceptable as dependencies of OpenStack projects,\nexternal libraries (produced and published by 3rd-party developers)\nmust be licensed under an OSI-approved license that does not restrict\ndistribution of the consuming project. The list of acceptable licenses\nincludes ASLv2, BSD (both forms), MIT, PSF, LGPL, ISC, and MPL.\nLicenses considered incompatible with this requirement include GPLv2,\nGPLv3, and AGPL.\n```\n\nTherefore, the project fits nicely into these requirements. The project\nin question is licensed under MIT, as we can see in [2], in the right\ncorner part of the page, and also in the LICENSE file at the root of\nthe project.\n\n5 - Is the library already packaged in the distros we target (Ubuntu\nlatest / Fedora latest)?\n\nIt is installed via PIP; PIP is available in major distros. Therefore,\nit is possible to use this library in all of them by issues the command\n`pip install DateTimeRange`.\n\n6 - Is the function of this library already covered by other libraries\nin global-requirements.txt?\n\nNo. At least, I did not find it.\n\n7 - Is the library required for OpenStack project or related dev or\ninfrastructure setup? (Answer to this should be Yes, of course) Which?\n\nYes, it is. We have a new feature being developed ion Cloudkitty in\n[4], which needs this library to handle DateTime ranges.\n\n8 - If the library release is managed by the Openstack release process\ndoes it use the cycle-with-intermediary release type?\n\nIt is not managed by the OpenStack release process.\n\n9 - Do I need to update anything else?\n\nNo.\n\n[1] https://github.com/thombashi/DateTimeRange/releases\n[2] https://github.com/thombashi/DateTimeRange\n[3] https://governance.openstack.org/tc/reference/licensing.html\n[4] https://review.opendev.org/c/openstack/cloudkitty/+/799207\n\nChange-Id: I608331e28ef349399068f9b1fa16bd6a6c5779e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/71a8c3bd4f8d665d569fa94b113be475ccfb9879"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/71a8c3bd4f8d665d569fa94b113be475ccfb9879"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
