)]}'
{"id":"openstack%2Frequirements~799347","triplet_id":"openstack%2Frequirements~master~Ia475c34b0f291f891fab34e1671e69d41831c35d","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Ia475c34b0f291f891fab34e1671e69d41831c35d","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-07-03 06:27:37.000000000","updated":"2021-07-03 22:18:04.000000000","submitted":"2021-07-03 22:16:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799347-openstack/requirements/constraints/noclob","meta_rev_id":"ae56097882a6b0074a103a9cfb86b65d5b8c785f","_number":799347,"virtual_id_number":799347,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-03 22:16:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-07-03 20:48:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-07-03 20:48:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-03 06:59:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-03 08:03:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-03 19:24:08.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"10c740c3ea3314aca6691170e6eb5012d545d44b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-03 06:27:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3bc5e8fe0fb4531a7221204d5489eb0f4606488","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 06:59:44.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/98ea598bdb9a4480a1d303b20e149f0c : FAILURE in 12m 35s","accounts_in_message":[],"_revision_number":1},{"id":"010829d8c214db7cbe38592d0b4f4cefddcc937f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 08:03:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb020fb8ac1346a891ad401378f17727 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63eacad69e994188b3d8733c04df0543 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d17847264a47443da7b17a159769dc69 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/daeae5fc7c6643aca389f126ebcc4d5e : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d9a66e8f7c5405b8426dd049c7b8250 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2a6bd504c9324c2c8cb4e3d34918b12f : FAILURE in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6279f89489ce4c949489a3755f794cc4 : SUCCESS in 3m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b3e4a4647e064873ab566a6fa7e5d25b : FAILURE in 5m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/60778428aba04c79a2693ba77a98ca35 : FAILURE in 5m 08s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/4172a12754a545c68960f6b00e5626ac : SUCCESS in 6m 39s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/394d1dde18c14b4c9813f6a660d80140 : SUCCESS in 10m 16s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/446e626cf8ac4e70a371c42dbdb3f3bd : SUCCESS in 9m 11s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/de60ab3fd9584fc89c3849d46a059677 : SUCCESS in 12m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ab802dc42d4f43abb38bdd0dc98ffe8c : FAILURE in 7m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5f4e99bb52e44e09bb8d650c6a4f4463 : SUCCESS in 8m 18s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5fa6b487f4834b2eaec9d17262c3b5a6 : FAILURE in 15m 24s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e39bb851d7724d0b9b2f54ba52f3eb25 : SUCCESS in 8m 37s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8470e48b479f4c9dae09a4075f82680b : SUCCESS in 5m 44s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ce471518514644cb94739b07be2fea1e : SUCCESS in 25m 04s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/362bec7a227c4d758615d7a59f2ea821 : FAILURE in 12m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/36f83ffa6a61416689cd2fca0452abff : SUCCESS in 4m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d9688624be0349538763c7ed837e0140 : FAILURE in 19m 22s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/198b5a3e3c024782a435cc90e51cad62 : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/64ee5d53451e47d6935872a6fda9abce : SUCCESS in 6m 43s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3bebed800a0e4242a36fdc099b4969b6 : SUCCESS in 12m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd94fa6aced94aca8cdb85247a562d8b : SUCCESS in 56m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bd0c634a803144ae85db79e5031dd9b1 : SUCCESS in 1h 15m 39s","accounts_in_message":[],"_revision_number":1},{"id":"c0568237ee59f574c619720d646165dd528a0109","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-03 19:24:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e5bace86bda0dd8086dfc4332f4d91984ef2cb3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-03 19:30:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c92f77b98f0119577657c213f9fc7ff17413074d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 19:46:31.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4926c70e264f41489911fa4bbf0b71e7 : FAILURE in 14m 46s","accounts_in_message":[],"_revision_number":3},{"id":"5c1dc14afa46ff439f623061cd7d88b4d078b3ef","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-03 20:28:21.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":3},{"id":"43c84d7de33f6bfc1770094a198e2a4131c0bc5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 20:44:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39536e5de71c4ba08dd2a7c08ad6859e : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bb8cd2ec48f4d98b15ee317ae75e653 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8add936269ed47d887998af62023df4b : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a98fefb9082486291566b0ff95e2732 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e32ca61a10144b72b7480b89197eee31 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d54addc626cd4ed2b23857ae18bc33fd : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c70b23b096b648cca8ae1cbaa84b332a : SUCCESS in 4m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1043886512ba45558884d8c02d9e7ec0 : SUCCESS in 16m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f727320c3e5c44409d3b1fb48dc9ac2c : SUCCESS in 10m 46s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b86b1b5f4d624ae389996b0d31815515 : SUCCESS in 6m 03s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/dbde4fcc7ca04c2d8f206b72c827fa12 : SUCCESS in 10m 12s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/afc15cd47850491aadb000d17292fb78 : SUCCESS in 6m 24s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/bacdf23288f64d268a893c3b02b5e744 : SUCCESS in 8m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ef4fce576b9346319677410e54a010cf : SUCCESS in 8m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/820a5943bd2a408b9084faee83c33925 : SUCCESS in 9m 04s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/eb126cf86fce4a93ab055cb3a4952f31 : SUCCESS in 41m 04s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/aeb84e0790c04164ad4ca112c3f56b07 : SUCCESS in 11m 45s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/783355ad1a554dadb46f94a50109a10f : SUCCESS in 6m 40s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ba49b54314e1465ba7baa30042d980f3 : SUCCESS in 34m 04s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/178471396fd543f6be8844af83d91c52 : SUCCESS in 12m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0a90502c69d645b48fe62114265537ad : SUCCESS in 5m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a8e2602669dc4482a52cff096a68ed4c : SUCCESS in 31m 33s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d9970d4e29ef461fb1ae6bcf40343f6a : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4fcaac0e0f1540e7a1aa05845e975a61 : SUCCESS in 6m 59s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0001328d50e3413db77db415ff9214fe : SUCCESS in 13m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a565a282f46647f6ab6bafa4db508075 : SUCCESS in 53m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/987dca84572142758e24dae7be6cb124 : SUCCESS in 1h 13m 04s","accounts_in_message":[],"_revision_number":3},{"id":"fb266c8ad5f3682aa1b6c8633f4fd1b6283325f1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-03 20:48:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"aedad154f92fc33cc86bdea212e3af4f2441f5b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 20:49:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5abc942d2a26572ee0bd140d189a5139f7797568","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 22:16:51.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24aac5c28dd943b88f274a420f5a4da0 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbd68b4d941241f59d08b9c698bb4473 : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90d2ebc0544d479bbfef69c611800683 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ee559a4620e434fb435d9ff73c8646e : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/46e97956e96e47c8a489e00bd23a656e : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/93906115b0084611904200f8e4e6ec99 : SUCCESS in 4m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bc07307083e94a9fba0e938627c159f9 : SUCCESS in 8m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b8af5db8b4054db8987a8b4ad1e94b89 : SUCCESS in 10m 05s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ced2c5e76d5a42c6b7a49a2cdf12dc34 : SUCCESS in 9m 11s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e91f780c0bd044c291bfc284ae5d0194 : SUCCESS in 5m 15s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4ebc41640aee4c45b322834c605ba01f : SUCCESS in 11m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4a1fe1dbc8114836baacf8425d81c6c6 : SUCCESS in 7m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/85d056aec64c49c0b8d9efbe22e1fea4 : SUCCESS in 9m 44s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4077ad3dc5634b69a7e3d1228fe17caa : SUCCESS in 15m 28s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c53287a95e9943668a9e249565a2a0a1 : SUCCESS in 9m 21s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/20c006037ff14d7d8b1285f4ec0a42da : SUCCESS in 7m 30s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1632c124685d4e14ae2e74eff290a9d8 : SUCCESS in 21m 49s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/01b5b0159c994ea7999c5548122736a4 : SUCCESS in 11m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b71c713e6e7b4c6fbac2bfe8add7d1a2 : SUCCESS in 6m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b9eec3bd66f44be783707561c1271ff6 : SUCCESS in 23m 29s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/ae961f549f994b86aace07d576dd3f12 : SUCCESS in 7m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6e5cb15e98cf4bcda15f569a92a1a69b : SUCCESS in 6m 23s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6e669e4f9e244149b02b639dbd6c8575 : SUCCESS in 12m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9eb2c73c98a74cfeb2144d99e141dbcf : SUCCESS in 39m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e4657e3703804d068e303dde533fd958 : SUCCESS in 1h 26m 14s","accounts_in_message":[],"_revision_number":3},{"id":"259e012c8c7ad80db45c86ae1a19da84dd76929c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 22:16:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ae56097882a6b0074a103a9cfb86b65d5b8c785f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 22:18:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f84c89d494448f7825a6ce422a4d40a : SUCCESS in 57s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a37962bcc6dbb3d9dc8659dfc8b0a699afe6fba9","revisions":{"0b2385361f778789f77c01ab063e891e94983a2f":{"kind":"REWORK","_number":1,"created":"2021-07-03 06:27:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/47/799347/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/799347/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/799347/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/799347/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/799347/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/799347/1"}}},"commit":{"parents":[{"commit":"5c4901e38e5ee51cd28cf6e42ca904f19a001f47","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c4901e38e5ee51cd28cf6e42ca904f19a001f47"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-03 06:27:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-03 06:27:35.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ia475c34b0f291f891fab34e1671e69d41831c35d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b2385361f778789f77c01ab063e891e94983a2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b2385361f778789f77c01ab063e891e94983a2f"}]},"branch":"refs/heads/master"},"1e812329350e0ed168f2f5d82de9395a1144cdff":{"kind":"REWORK","_number":2,"created":"2021-07-03 19:24:08.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/47/799347/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/799347/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/799347/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/799347/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/799347/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/799347/2"}}},"commit":{"parents":[{"commit":"5c4901e38e5ee51cd28cf6e42ca904f19a001f47","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c4901e38e5ee51cd28cf6e42ca904f19a001f47"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-03 06:27:35.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-07-03 19:23:57.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ia475c34b0f291f891fab34e1671e69d41831c35d\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1e812329350e0ed168f2f5d82de9395a1144cdff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1e812329350e0ed168f2f5d82de9395a1144cdff"}]},"branch":"refs/heads/master"},"a37962bcc6dbb3d9dc8659dfc8b0a699afe6fba9":{"kind":"REWORK","_number":3,"created":"2021-07-03 19:30:12.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/47/799347/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/799347/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/799347/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/799347/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/799347/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/799347/3"}}},"commit":{"parents":[{"commit":"5c4901e38e5ee51cd28cf6e42ca904f19a001f47","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c4901e38e5ee51cd28cf6e42ca904f19a001f47"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-03 06:27:35.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-07-03 19:30:07.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ia475c34b0f291f891fab34e1671e69d41831c35d\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a37962bcc6dbb3d9dc8659dfc8b0a699afe6fba9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a37962bcc6dbb3d9dc8659dfc8b0a699afe6fba9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
