)]}'
{"id":"openstack%2Frequirements~799452","triplet_id":"openstack%2Frequirements~master~Ib26f6b8e139d3edce0bdaaae7cee4d00d6ed6391","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib26f6b8e139d3edce0bdaaae7cee4d00d6ed6391","subject":"update constraint for osc-lib to new release 2.4.1","status":"MERGED","created":"2021-07-05 10:09:36.000000000","updated":"2021-07-05 21:38:59.000000000","submitted":"2021-07-05 21:37:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799452-new-release","meta_rev_id":"4a9d5380bc6cc1ba2b160395f5ea5a59608ca268","_number":799452,"virtual_id_number":799452,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-05 21:37:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-07-05 20:17:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2021-07-05 13:56:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2021-07-05 13:58:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":2,"date":"2021-07-05 15:04:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-07-05 20:17:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-05 10:31:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-05 11:37:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-05 13:56:09.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2021-07-05 13:57:44.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2021-07-05 13:57:44.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-07-05 13:57:44.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"970a64d1cac8ad8aee74cf99de68e318aa483d0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-05 10:09:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0064460ebf2636fa94c7bccd3459b58edbd4185b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 10:31:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d0d193ed6d5b4ee9bfd048b4d9480b2a : FAILURE in 20m 22s","accounts_in_message":[],"_revision_number":1},{"id":"a155630a9204c7a7bb10b89f0b9a6ce016439f65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 11:37:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2183e3926ac541b2b25beb4bc50d70a3 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dbbbc7972e44d9aa881227bbcd7ca69 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d55e73322ae34d86a0eaa4c503df0de9 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d422a1b9a664876935b1b3bfd5f3be3 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6bbd90526c145b390d18421c2e946ff : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8c2d49a3717b4f6892dbf398f635a2c3 : SUCCESS in 4m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a319b687fecd483ba5df04a5f2efdccc : SUCCESS in 4m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/166ee36341c14274a72ade3a8c1b20e4 : SUCCESS in 10m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c0526ad0fb0f4f4dbcdd22d246b1f05b : SUCCESS in 12m 11s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ca034562e9f34e119288e0e012b73187 : SUCCESS in 6m 06s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d5ce010dfd2f4740bc9d4b1e077573e7 : SUCCESS in 10m 57s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8c54c1787d6e468a90cf8e279685a60d : SUCCESS in 6m 12s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1a3b42299d0e4a519b0dbe2050fb001d : SUCCESS in 10m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/49428f69148e4c1c97c78f7cea03287c : SUCCESS in 9m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7ab0351ce2cc41ee9e17e243f2a11516 : SUCCESS in 9m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8a0edcf79aa94a09980fab14c71d2e37 : SUCCESS in 13m 24s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/202940b82a074d84b5aba813f43378ca : SUCCESS in 10m 48s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/74dad93c03a642e7a4bd1cf0b351d78c : SUCCESS in 6m 21s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/082c2c5b0a134381be845ad9f09dd35b : SUCCESS in 27m 51s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0ed7e1ee43ad49dc8015edc122f0fe78 : SUCCESS in 13m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1ab7193370d64f439cd2b8b2c0493521 : SUCCESS in 4m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2be8e616af5d42649894ee2c8a498b56 : SUCCESS in 21m 49s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/589e0ed0eb92467f8ba23af6fdc020ed : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7f98be8e9c404902b44e16fadb091241 : SUCCESS in 7m 27s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6ed64d3476404f87803544205cabb549 : SUCCESS in 13m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/781e49ef42b24a979093a1fa5bd2f7cc : SUCCESS in 49m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76894148f6c141d29d139357a1ca589c : SUCCESS in 1h 23m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ea17fb17eec7ac197a5de749cdfd524d80171847","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-07-05 13:56:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cd86fae1114452351dae12cb370d209de3d3a0b0","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-07-05 13:58:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7351307c605a46091e0568f5fb8d133ad9053066","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-07-05 15:04:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e08e06de2fa2522a70950596363d8cb8c0bf32cc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-05 20:17:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"048c7a214855bf427e13fac6c4ab85b40d13a8cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 20:17:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"35286c30b14b31b282d98c7df706dd29c5de11ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 21:37:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63373fbfc5b3411d80bba083b34f6754 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bd842ab44074fbaa7c35ed06d5c67dd : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b0ebb2d66264f83b6f214f561a61c04 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f6e471fd69b4abebb5aa862bc653702 : SUCCESS in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c3cbc240f64d49188e163e8173e94495 : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b974e0b047cb490995d0d9b326ae394b : SUCCESS in 4m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3f2d3cfcb766443fb711c5d527ea7411 : SUCCESS in 12m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3998471474fa49d1bcbdac4e2d155bfa : SUCCESS in 11m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f78767f629304134a43feb72922ac5b0 : SUCCESS in 10m 48s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2d6f53401c5549ecbda9bbfa3e72dcc4 : SUCCESS in 6m 25s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/569c8e1e20de40a4b4aeff908d7bf098 : SUCCESS in 11m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c485329034614bc5bb5ba20f3ebbb35d : SUCCESS in 8m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6c2c728717bf40059ee071eb98338c8c : SUCCESS in 9m 55s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d369a424f4ae49ada2a6c430048bf05b : SUCCESS in 15m 07s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ba4fd9a85f714df9848bfc9ac391ad2d : SUCCESS in 9m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/37ad92499d634485a248dde0f8176aa4 : SUCCESS in 5m 29s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/54955015860f449f80a55eb23db5e9a9 : SUCCESS in 25m 55s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/067d3e38f7f74d06920116ab758f0f8e : SUCCESS in 13m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dc396deb323b42a29cb9bba526ed2464 : SUCCESS in 4m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/694559323bbd4a22b936ea057b40e0a5 : SUCCESS in 20m 57s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9f6c769ec5794f03990cc4e4a2c9c64f : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ae2801e3f45748678a4a4cc3dd81b1df : SUCCESS in 6m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/287e577879224c3384959ea0226cb05d : SUCCESS in 12m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1344768c84184fb994c06a3b44952c64 : SUCCESS in 47m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/135db4b17825404cb310ea9f11d5c71f : SUCCESS in 1h 16m 08s","accounts_in_message":[],"_revision_number":1},{"id":"fa128e3ce2a0d7ac20606fe602bc51064119aac7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 21:37:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4a9d5380bc6cc1ba2b160395f5ea5a59608ca268","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 21:38:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85497373789646ada3c691171a43a19b : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c9f6aafc436655bdd3940aa9ae583adafa7b5606","revisions":{"c9f6aafc436655bdd3940aa9ae583adafa7b5606":{"kind":"REWORK","_number":1,"created":"2021-07-05 10:09:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/799452/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/799452/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/799452/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/799452/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/799452/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/799452/1"}}},"commit":{"parents":[{"commit":"08fc133101542dd44a38da9c4a241f6c01f2f669","subject":"Merge \"update constraint for oslo.rootwrap to new release 6.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/08fc133101542dd44a38da9c4a241f6c01f2f669"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-05 10:09:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-05 10:09:33.000000000","tz":0},"subject":"update constraint for osc-lib to new release 2.4.1","message":"update constraint for osc-lib to new release 2.4.1\n\nmeta: version: 2.4.1\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: James Page \u003cjames.page@ubuntu.com\u003e\nmeta: release:Commit: yatin \u003cykarel@redhat.com\u003e\nmeta: release:Change-Id: I97f38b96bce39d66072128141470679d2c03cc1e\nmeta: release:Code-Review+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Alfredo Moralejo \u003camoralej@redhat.com\u003e\nmeta: release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Code-Review+1: yatin \u003cykarel@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ib26f6b8e139d3edce0bdaaae7cee4d00d6ed6391\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c9f6aafc436655bdd3940aa9ae583adafa7b5606"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c9f6aafc436655bdd3940aa9ae583adafa7b5606"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
