)]}'
{"id":"openstack%2Frequirements~799741","triplet_id":"openstack%2Frequirements~master~I403dca5f27c2508f2fb1a91295b8e2a66628841d","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I403dca5f27c2508f2fb1a91295b8e2a66628841d","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-07-07 06:23:08.000000000","updated":"2021-07-12 21:14:00.000000000","submitted":"2021-07-12 21:12:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799741-openstack/requirements/constraints","meta_rev_id":"b9e4e69e430d4bf385d76a2d7fe08a1b0c4b10e2","_number":799741,"virtual_id_number":799741,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-12 21:12:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-07-12 19:50:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-07-12 19:50:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-07 06:38:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-07 08:11:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-12 15:37:47.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"8ce42f283a112d1209a7249a8081c9b6f6ff4741","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-07 06:23:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7f3fefa54c8e37439c06f47d7e5108393413837","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 06:38:52.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c6c741704a554517960d751973a1db98 : FAILURE in 13m 33s","accounts_in_message":[],"_revision_number":1},{"id":"49f714262fe4ee13728c30532be63349ee1b26cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 08:11:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eab2176341a94d40a50fdd566a31574e : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c38ebc425cdd436493f1682177784795 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0c21d672680492cb7ef1d203eadf22f : SUCCESS in 8m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd029f40d6334a5eb76060cea9c7996a : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef693f72e5d0437fac7316dc6ea47c4c : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d6be7f85f2534dfd8ee94243c9a953b1 : FAILURE in 6m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/75bc72d14dd04c4db1ac4241ea58e38e : SUCCESS in 4m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/70df9b6d1fec4472be1bf0f96a15da82 : FAILURE in 9m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9b38c1d512c04e77ae127452c50325a0 : FAILURE in 8m 53s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/9ba4dffd8f9e494298c645b2619ef903 : SUCCESS in 7m 35s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e26b52e678594c608e2736a36bfbd1b5 : SUCCESS in 11m 24s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/6e0b2f962c754a489aefd0068c70afb5 : SUCCESS in 6m 45s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/17549da1a3ca4cc9865d9acba74e62fd : SUCCESS in 10m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/aa00da3ad08a4d70abac028214d46df7 : FAILURE in 8m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b93ca9d53f8a4edf93d3849344fdf50f : SUCCESS in 11m 33s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b7aba5f04e5643c2bb79687fbed71f80 : FAILURE in 16m 42s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/10ca00b8a33b4d1e8005510f3cc9a27b : SUCCESS in 10m 33s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/20bd1670b51e4126b8b8d8abd29c7164 : SUCCESS in 7m 38s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/bb257ad0cd174b8ea5f0344e83ff3d86 : SUCCESS in 25m 16s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a323583af5e04f99bd9d432dba43705f : FAILURE in 13m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b5b84c7289e4441d931159b9eb5b426c : SUCCESS in 7m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4b7f47428278466abb63d1d1728515a1 : FAILURE in 18m 00s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0c0050ffb5b44fcab334406497e5c8c5 : SUCCESS in 7m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a98109006d8a479491b25b62de6df229 : SUCCESS in 8m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/92407419508a42e88688345a60ba0852 : SUCCESS in 15m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/155a21bfcd5e45dcbec783acfa7198f9 : SUCCESS in 45m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/396cc2e2e1b74548aa539af96c4ce031 : SUCCESS in 1h 25m 28s","accounts_in_message":[],"_revision_number":1},{"id":"15a852ac4ffd4130e0e128a2c584e53c462fa15c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-08 06:22:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"63ad5ad27d2d0952699300685afcecfc3539e486","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 06:37:41.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8dfe434aeac04fbe870833fa27f31fd9 : FAILURE in 10m 52s","accounts_in_message":[],"_revision_number":2},{"id":"6adbf2ba388cae5d693e5d450184684c69009749","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 08:11:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e6748076c744b55a4082464947e8399 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fe0ca8ba2c945f5acc521ee2bfe45a2 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0d11efe3a494ba69cc633500633e673 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a2b527dbd3046df9abec3fe2093d778 : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4e1247f10cf436681da4297f435705e : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f290a7dc700a4dd4ab3cdceca10fa37d : FAILURE in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1f89c4aa456440ea8cfd49cf02998266 : SUCCESS in 5m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5935615c000f42009af3ae17a0f3b588 : FAILURE in 6m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4780433a0d6740aeb1b06a68fd83c295 : FAILURE in 6m 00s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/4cf3ff6e039b42618faafec60ce5c12c : SUCCESS in 6m 47s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f7db83ef826f475182a6c45c95cd4c3e : SUCCESS in 26m 42s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/dbd44cb4d2b8472db4da4ff19d8814af : SUCCESS in 20m 24s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7720baffd9454aa9b2d59fe2861e1034 : SUCCESS in 21m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c592c35fede54eb1ad358b5617e9bd27 : FAILURE in 7m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/28c8c3e545bb481abf766633fc8c3f05 : SUCCESS in 8m 41s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/af491966d7ba48d4a0f1c0466db978c0 : FAILURE in 13m 59s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/eb7d2237393b4eb7b953e7933df6dc20 : SUCCESS in 9m 59s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7325b43bda47464cb1ecffccc03b4d8c : SUCCESS in 5m 49s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d321beed89924f109b15f34adf3d21d5 : SUCCESS in 21m 49s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e587cd2af44f4670b640e2e8c1b3db48 : FAILURE in 12m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f6afd1cb04e34f2a8b001c3e46c27836 : SUCCESS in 5m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/882af541c3ba429598778533f915c73b : FAILURE in 18m 43s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3dabf1ad9beb4a798b13db4e0250be62 : SUCCESS in 6m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5aa207a6abd74ede9d6ea68a2f493f98 : SUCCESS in 21m 22s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/dde0111b26f6476ab2f43fa1b1a100ee : SUCCESS in 14m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ca3747c353b4e1db5394ddc0a674b52 : SUCCESS in 46m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4542c3b8289348a7a8c9069a0088404f : SUCCESS in 1h 22m 11s","accounts_in_message":[],"_revision_number":2},{"id":"fed0e6209d600cf32df0e67e3729b88afd02ee70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-10 06:20:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"24b3515f5f402953bb99616a6676fa91474da86a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 06:37:37.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7c485de1dbd64a0dba04b39491a050ee : FAILURE in 13m 50s","accounts_in_message":[],"_revision_number":3},{"id":"a92f2c9f661dcc9252bf26e83e0d9d18d24b9b92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 07:42:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06b8059b622f4ec6add695b5f8c64b6c : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4368336118554f3d9873b4ddb0379731 : SUCCESS in 9m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4e1f6c8f8ee4563aaa44e0ef469216c : SUCCESS in 10m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab351e03a0f7458f96d72f0620b1958d : SUCCESS in 6m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8774aa3b631248d59d3a8ebdb3ab9791 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/07b1110d2b8d4aed9f0a10512dc785dc : FAILURE in 5m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b378bd82455e4e97a90c47bdd90bad52 : SUCCESS in 6m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/af10eaead28a464c9829ef4ea2a08d4d : FAILURE in 8m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3c3659b306e04e6a8b0b39659ad37c1d : FAILURE in 7m 25s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/c1e0f246779e4ddca67a813b9f8430a4 : SUCCESS in 8m 00s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0b91e8043e6743598e29dec1e92ebc6b : SUCCESS in 11m 34s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8f0da9c2549d49c8a351f8ef98065917 : SUCCESS in 7m 53s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/31b41f13b6ae4ec48ddea31448948c69 : SUCCESS in 12m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c2386dfb15e8454f96f8b214af5f8e6c : FAILURE in 12m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/75c3c450a0754b04b24c6552b2e91a91 : SUCCESS in 12m 33s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7e1d014b411f4ad79236523f4d0eb4d2 : FAILURE in 21m 27s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8d2b9f932a4546adb0df286f79faed96 : SUCCESS in 14m 01s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/53af2d5c79a147a1a01fcf57f3872f44 : SUCCESS in 7m 24s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3020c831f77c4621a8594d94db5f82d0 : SUCCESS in 26m 35s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1441add071964eafa925b6f65e53c824 : FAILURE in 13m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/051866fa26a14d6ca8d72bf3e3afb5bb : SUCCESS in 5m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f38b2cf793114fa291da2dfef579144c : FAILURE in 20m 07s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b83ebb554a1642c99bf0e5bcdb07e9e4 : SUCCESS in 7m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/09d68e76fd5447e0b70c97b62ba692eb : SUCCESS in 9m 28s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d0be1d4ffdd14599a91a703fdf439c83 : SUCCESS in 14m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/97c923717215402ba44946265fc2b766 : SUCCESS in 1h 06m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8b19472a69254ac88bbde2c4da84dd6f : SUCCESS in 1h 17m 18s","accounts_in_message":[],"_revision_number":3},{"id":"b7b381f8134591cf8a04bf1aea5e69494aa1911e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-11 06:29:39.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f147f0b9ff1cb6c043a1223e5d783d2d31f82a18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-11 06:44:44.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4a30bbaae0bb4cdbaa68374b4f4763c2 : FAILURE in 10m 17s","accounts_in_message":[],"_revision_number":4},{"id":"03226f436b5a1a24bd29ef3050b1d8c945203d41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-11 07:44:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfc2757c6cbb41f190c773a2415f0d82 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7104e70329343e4988f0ed77771b89b : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/baa37f1e80e349acb361a268dcfdff1a : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0382a0899fee4bb3aa091ec799173a32 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f02834da90444ec9ae97a69339adb5f8 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/421ad1ca742a40118d46505cd46674ab : FAILURE in 6m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7bcf04e23f0b47f7a88ab8b6f0bcf02e : SUCCESS in 5m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/23caae120a364f15a0c2b22ba6b48325 : FAILURE in 7m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e61636c84f9e49afa52b36b5b6db1992 : FAILURE in 6m 13s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/1c1abf76db5b4e509ebefd2cd18b2bc1 : SUCCESS in 8m 21s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ee810f4105d74b759e05fd955d3a0f94 : SUCCESS in 11m 10s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2de1c192209b46a0bf43329f37e01093 : SUCCESS in 7m 06s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d0d9dcc6ed17496ba63a2efc6542ecfe : SUCCESS in 14m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/761f3c53e7414984af975e585ffd6c84 : FAILURE in 19m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/39157758cf38468190eb843eb7fae66a : SUCCESS in 8m 28s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1bbac90284094f8583d48f285d20e592 : FAILURE in 14m 01s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8b991fc8e895476ea9af5e325facd745 : SUCCESS in 9m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/bff1afd5687f4efcab0b895b6230e97a : SUCCESS in 6m 43s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d7075ed392a44a6aa1923a0334303df3 : SUCCESS in 25m 36s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/276f4d315bed400d9f39f6c43b84893a : FAILURE in 14m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4901506b2b824491a290769aac6a34c8 : SUCCESS in 5m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ebc4f19411c5497b8f30ea60409b8d4e : FAILURE in 17m 55s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/bd285a2ed5214fc08c3cfe7b499bc351 : SUCCESS in 11m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/72d6250c90744fce969297b7f87c2dac : SUCCESS in 6m 49s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e7a94b6cbed74d2c9699803b5ae68301 : SUCCESS in 13m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ffd23c6e7ed84a15b7c2a4bbfcb6a3af : SUCCESS in 45m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b1a7ca6d09374cfbb2984a8ede815e46 : SUCCESS in 1h 10m 59s","accounts_in_message":[],"_revision_number":4},{"id":"2d0f9dfa9a9f26093fe23d154740e45ff865c70b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-12 06:24:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1c11fdb7d68f247db59dd91f5f42dc41ee71e4c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 06:38:35.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b40d4e7860f04251a3237722d10584b4 : FAILURE in 9m 44s","accounts_in_message":[],"_revision_number":5},{"id":"9b33f57acbc9bc585d88a13a69352e7049f2ab18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 07:51:25.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e35e323b83a478fbd307dadaabdc4a2 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/367d86a12f8c41afb3aa2553a7a2b2be : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c29d75947094ef7bfa4055c3281353a : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/049a030d6bc24cf39d76b55479539fd6 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54dc2cb255094800b8e4dcd89e02f40e : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eb85b6281a624fddb1b7a883284b124e : FAILURE in 6m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dfd2cce2b58c4918ade6600b43a95a5f : SUCCESS in 5m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fb17eb8e749a4dde86c18be3e991843d : FAILURE in 7m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b34d7dc71acd45dcae9f1fdfcf31244a : FAILURE in 7m 58s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/54f0ebb1b8e546559676ab13eaf4b4ab : SUCCESS in 7m 21s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e62429f33c9740e3869b2cf0bf5bc9c5 : SUCCESS in 10m 54s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9813ac07ac7148bba794a0a16b823f1f : SUCCESS in 8m 11s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/259752634dad43b8ad045373fa52732f : SUCCESS in 11m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ed479688014742ed899854069c973b19 : FAILURE in 8m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b4594102b76b4c2b8e5982a22febb97a : SUCCESS in 8m 40s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7bab486d9e3a4eabb05235cef395432b : FAILURE in 13m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b87e114585c144efb0012cfed8f612c8 : SUCCESS in 9m 44s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/3913d80b59f74a7b9a701609c72b57d4 : SUCCESS in 6m 02s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b316b7c885b54c018b2b10e7a4b09738 : SUCCESS in 24m 34s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5ac0a61c9e354c71b3a9b12878c04156 : FAILURE in 11m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/60017e77370f4196a0f6370ccc3a8291 : SUCCESS in 4m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1f10f25345154e8ba337e63127090f83 : SUCCESS in 18m 31s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/55f8f9eef4184e87b9ddfc875d6a2ec0 : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/11bf8e6b9cbe48aa95eafdd60fefa414 : SUCCESS in 7m 28s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/aa405cbfa98a4437a75250f28680a2b8 : SUCCESS in 15m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/202bb01f79aa47dc99a6436423f71518 : SUCCESS in 45m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/69ccbed5be9b46ca90e1bdc4c378a13c : SUCCESS in 1h 10m 46s","accounts_in_message":[],"_revision_number":5},{"id":"6642c211aff9db6fd48babf35b37aa8c3dc5b7ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-12 15:37:47.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2e9fc3c64068043cc1f906117a966b02d46c3624","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 15:56:28.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3b35143865b54fd499bcac253cbc35e4 : FAILURE in 16m 41s","accounts_in_message":[],"_revision_number":6},{"id":"a40b46a6c161023df3c90dcc58e19334e476da34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 16:43:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/898d658ae91b445fb72d5858a5d5fce1 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/159ac7f7d0e347deba4c3436bb1287f7 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c6dddd6b751412cadcf094287d4ab95 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce3814381ec1473986f70b7a13e57097 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a241640ca9f43558bf33afe6237eb54 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a9e53cd8a6c349cb8c1e254e4048f943 : SUCCESS in 4m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/30cfb0f860e54c29a515f8dc99f4f0c1 : SUCCESS in 6m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1bc9f3b6b91240fcb70fdb69863a3d50 : SUCCESS in 10m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b09c5c3fb700494a912d65b207dadc32 : SUCCESS in 8m 21s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b82551b558db4933a81fa7f18c3b8003 : SUCCESS in 5m 42s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6468fd2c4c2f4273a68481725e2c5d49 : SUCCESS in 9m 53s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/478e8c5a05ba405dbf8eac7640cae81c : SUCCESS in 5m 49s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/341f850313124277b772449e749c2e5e : SUCCESS in 9m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/499b8946cba640899312d9c0b0da2c17 : SUCCESS in 8m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1a33551fb08049c39a08b1da33fdba82 : SUCCESS in 8m 40s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6d7e39d8bbaf46d69871c2bd3f566159 : SUCCESS in 13m 34s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/afe3e309c49842ba9923238744048365 : SUCCESS in 10m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/80d33e0d64cd4f3fae714e55161655a1 : SUCCESS in 7m 46s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/980f31d56b9c4029bb286f8185c066ad : SUCCESS in 21m 15s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/084dacf7411f4868b335a33fed6e40c1 : SUCCESS in 12m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cce12c3b014343129215655dc7fedc0e : SUCCESS in 4m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d6e3875fd7834a7b9ba7dcdb5603b9be : SUCCESS in 17m 57s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2fa526eac48c4e46a3e1c7051a67cd85 : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e1cf4b3ed721403090ca156a88d3e085 : SUCCESS in 7m 39s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b02cd93fd0cb4181995aa446eb603571 : SUCCESS in 20m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9da7db054b284633aeaf07150fe0d70b : SUCCESS in 44m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/893670919d464b35b5789887967147dd : SUCCESS in 1h 02m 28s","accounts_in_message":[],"_revision_number":6},{"id":"c84d855651b8487f4253ba293fbbeef34b328e9c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-12 19:50:54.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"f85a136b3fd3c138a340b473cf8e7c1b788ebf48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 19:51:15.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"04ab7afc11f431112d679d384218369f380b9939","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 21:12:41.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/102077c473f444cb9dd2b27f6be7dfd4 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f834cde325054017b099c680bbbac1e3 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec2b954702a64e50a6ffdd73f0c21cf0 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7856bc4576a74a30a9e42d87f2f010c8 : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d5a50dfc530349539132a1da26bb3154 : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/23d9e6c3cfef41c7b0ad11671357c6a9 : SUCCESS in 4m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/995027a6f62f463eaaede3e38d7bdfeb : SUCCESS in 9m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f72eceb326674c80b6e698be62bf8e91 : SUCCESS in 11m 06s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/58b977e51b2c41e29d842aa5955fa83e : SUCCESS in 10m 03s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/71251ec508d941fb9b5e0edfc666db0d : SUCCESS in 6m 01s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1e75f7da0e3f4feb92131cfda16f23bc : SUCCESS in 10m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f3d45914d30a420f9683723e7ba7d8b7 : SUCCESS in 8m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/af2ac292ccc54d70a192ed0f0703ba0a : SUCCESS in 9m 12s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/37b97dbc61dc4d20b1adc7cb68e75a0b : SUCCESS in 13m 29s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/61e0040301f84dbc9cdc322943bc006e : SUCCESS in 9m 01s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ec3bca04e3e546539f3ecc6420ebdb89 : SUCCESS in 5m 17s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f0b59e1bf74f40a4aa5760cf0d9df1a5 : SUCCESS in 23m 38s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e8b50467b5f0495fbfe928c339e620cd : SUCCESS in 13m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4d1642a404514263a3c665aa7002c14a : SUCCESS in 4m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c6df1cab406a4962a1b4277e1c5f5e85 : SUCCESS in 19m 44s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/21d2e78f1631446992464ba79eeb83ef : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8ea94be1050849e6a39d652d5b160a5c : SUCCESS in 7m 11s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2a471538d07a44d6a7934a30a7935373 : SUCCESS in 13m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/565a9d8063214f7c8f3be42a675b72c0 : SUCCESS in 46m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/de6d915a05d04492985bad1aeb1c1cbe : SUCCESS in 1h 17m 37s","accounts_in_message":[],"_revision_number":6},{"id":"fda82028b14d0fdab37e154d7872cca917f54ec6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 21:12:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"b9e4e69e430d4bf385d76a2d7fe08a1b0c4b10e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 21:14:00.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad126ac8a5814b8fa07e2216e397a392 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"cfe3356a9e84e2c1cdc9753489f9ba6663bf9229","revisions":{"36c81146917903f9295bba4b31206cfb2d50a607":{"kind":"REWORK","_number":1,"created":"2021-07-07 06:23:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/41/799741/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/799741/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/799741/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/799741/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/799741/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/799741/1"}}},"commit":{"parents":[{"commit":"c9f6aafc436655bdd3940aa9ae583adafa7b5606","subject":"update constraint for osc-lib to new release 2.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c9f6aafc436655bdd3940aa9ae583adafa7b5606"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-07 06:23:05.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-07 06:23:05.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I403dca5f27c2508f2fb1a91295b8e2a66628841d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36c81146917903f9295bba4b31206cfb2d50a607"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36c81146917903f9295bba4b31206cfb2d50a607"}]},"branch":"refs/heads/master"},"a1541e306522c4b2a2ab9f9de2dbabff0e408fc3":{"kind":"REWORK","_number":2,"created":"2021-07-08 06:22:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/41/799741/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/799741/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/799741/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/799741/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/799741/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/799741/2"}}},"commit":{"parents":[{"commit":"c9f6aafc436655bdd3940aa9ae583adafa7b5606","subject":"update constraint for osc-lib to new release 2.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c9f6aafc436655bdd3940aa9ae583adafa7b5606"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-08 06:22:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-08 06:22:46.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I403dca5f27c2508f2fb1a91295b8e2a66628841d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1541e306522c4b2a2ab9f9de2dbabff0e408fc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1541e306522c4b2a2ab9f9de2dbabff0e408fc3"}]},"branch":"refs/heads/master"},"342973bfe82a6abe9591fac9086b123294498ad8":{"kind":"REWORK","_number":3,"created":"2021-07-10 06:20:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/41/799741/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/799741/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/799741/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/799741/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/799741/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/799741/3"}}},"commit":{"parents":[{"commit":"1666f810c6428cb3924ee8091564ce1db37a93ef","subject":"Merge \"Deprecate os_panko role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1666f810c6428cb3924ee8091564ce1db37a93ef"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-10 06:20:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-10 06:20:48.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I403dca5f27c2508f2fb1a91295b8e2a66628841d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/342973bfe82a6abe9591fac9086b123294498ad8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/342973bfe82a6abe9591fac9086b123294498ad8"}]},"branch":"refs/heads/master"},"ac43fd07acff3e0603f8ca94e0f8c0631eb092af":{"kind":"REWORK","_number":4,"created":"2021-07-11 06:29:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/41/799741/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/799741/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/799741/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/799741/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/799741/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/799741/4"}}},"commit":{"parents":[{"commit":"1666f810c6428cb3924ee8091564ce1db37a93ef","subject":"Merge \"Deprecate os_panko role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1666f810c6428cb3924ee8091564ce1db37a93ef"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-11 06:29:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-11 06:29:35.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I403dca5f27c2508f2fb1a91295b8e2a66628841d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac43fd07acff3e0603f8ca94e0f8c0631eb092af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac43fd07acff3e0603f8ca94e0f8c0631eb092af"}]},"branch":"refs/heads/master"},"5c95f7d02e7a7e653be4a0ff551d3da5d93567a8":{"kind":"REWORK","_number":5,"created":"2021-07-12 06:24:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/41/799741/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/799741/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/799741/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/799741/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/799741/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/799741/5"}}},"commit":{"parents":[{"commit":"1666f810c6428cb3924ee8091564ce1db37a93ef","subject":"Merge \"Deprecate os_panko role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1666f810c6428cb3924ee8091564ce1db37a93ef"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-12 06:24:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-12 06:24:07.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I403dca5f27c2508f2fb1a91295b8e2a66628841d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c95f7d02e7a7e653be4a0ff551d3da5d93567a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c95f7d02e7a7e653be4a0ff551d3da5d93567a8"}]},"branch":"refs/heads/master"},"cfe3356a9e84e2c1cdc9753489f9ba6663bf9229":{"kind":"REWORK","_number":6,"created":"2021-07-12 15:37:47.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/41/799741/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/799741/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/799741/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/799741/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/799741/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/799741/6"}}},"commit":{"parents":[{"commit":"1666f810c6428cb3924ee8091564ce1db37a93ef","subject":"Merge \"Deprecate os_panko role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1666f810c6428cb3924ee8091564ce1db37a93ef"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-12 06:24:07.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-07-12 15:37:35.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I403dca5f27c2508f2fb1a91295b8e2a66628841d\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cfe3356a9e84e2c1cdc9753489f9ba6663bf9229"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cfe3356a9e84e2c1cdc9753489f9ba6663bf9229"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
