)]}'
{"id":"openstack%2Frequirements~800448","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I75f38cab25ad5f82b25eff2c61cbc243a37c8301","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"I75f38cab25ad5f82b25eff2c61cbc243a37c8301","subject":"update constraint for neutron to new release 16.4.0","status":"MERGED","created":"2021-07-12 10:12:56.000000000","updated":"2021-07-12 16:00:31.000000000","submitted":"2021-07-12 16:00:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800448-new-release","meta_rev_id":"5b758989e4152a20877f0902b88d7c900f72f8fd","_number":800448,"virtual_id_number":800448,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-12 16:00:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-07-12 14:25:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-07-12 10:23:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-07-12 14:25:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-12 10:23:28.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-07-12 11:30:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-12 14:25:20.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"161da5d7acac58f20797ae8daec60d3fa3a68e01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-12 10:12:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ef43063cbc1e950c5f22a86f1cdea2caca8599e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-07-12 10:23:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a71c41059f8158689695d5045b334aa4b62079b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 11:30:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3400752943fd40e287605f99d03cfc28 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b209ba6be4c42f38405094cc4827a4d : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6cbe1c58b2bf4348910a068073e231e0 : SUCCESS in 5m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4238cbbb1c084772991e08eb94af21ad : SUCCESS in 4m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4b8123a5e4fb4a02a277b8c16c31ec2e : SUCCESS in 5m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d0c55287da04402fa530f44fafdd988c : SUCCESS in 13m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dd876506c9cc40bbb4805c42f01a3547 : SUCCESS in 9m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4816dbd9f6bf4d949ef47b092eda7c07 : SUCCESS in 5m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8595a2b176254392869e36ae4d1a7368 : SUCCESS in 9m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a204fe6030194d0baea12c1e43f54cb4 : SUCCESS in 9m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1b5b168805e447b19bb2d8cb0544fa6c : SUCCESS in 19m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9b2ca4d319d04dba8fa5ec00a3f9e7dc : SUCCESS in 21m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f3bfa3ee3e044a93852c03972829fc0a : SUCCESS in 14m 20s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/eb39cb9b48444f71a4ebcee6aabc3efe : SUCCESS in 13m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/76fc6fec90814173bebd55147e1ee2ed : SUCCESS in 17m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d3c2fbe79e0342c2819716e29d1481de : SUCCESS in 5m 51s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/a2ca51c784eb4ae99ca2504c265105f2 : SUCCESS in 7m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9e34bfdd3567451fa0490ba4aa3714d6 : SUCCESS in 7m 57s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ac6bcabc9b274aa4ae0a14f64b93509d : SUCCESS in 10m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8a062ca6c0964369b92fb1e9eeeb14b5 : SUCCESS in 12m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5eb555e7c8094cf6be9812397fb1c8c8 : SUCCESS in 50m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3b5f6e41e90e4dbda541f218ac9bd027 : SUCCESS in 1h 05m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc992b60733e4e76a0500cd9559d117b : SUCCESS in 1h 15m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/3be5872c0d3c49fa9a1aecfc2352deef : SUCCESS in 1h 05m 10s","accounts_in_message":[],"_revision_number":1},{"id":"3ff1dd0e119040c531b156b7936a0d53ff1dd1cf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-12 14:25:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ce1f48d64aa4be02edf4936f1cc45da46962b7ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 14:25:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d845b59c4ae868cc4d7661bf012bb5171e643708","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 16:00:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/139ea101e7fd4ef5ae3fd444350f16b0 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a9643b1502f49238f6ee811b9d7aee7 : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/72fc97f459134821962fc584b61c9264 : SUCCESS in 4m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1062346f06b54d7eae6ef50559ab9c6c : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/98d542542c1d4a0f9d2c32309d472d4e : SUCCESS in 3m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d01803e7c17b4d8f90f7d2c4ad97d108 : SUCCESS in 10m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3ac6bf601d6444e9a89acb9f8d398e1f : SUCCESS in 11m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2e82a9fd5b2043bfbd39646e7ce988c6 : SUCCESS in 5m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3fc586aee31944b1a263ec0895425573 : SUCCESS in 12m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6c4f1cb37d8f4371872369d1072a579e : SUCCESS in 11m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/baab5c0cba6f4301935242135f09f613 : SUCCESS in 14m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cbaccae755e44bc5ace29d574cd62203 : SUCCESS in 26m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c7b6a3def77f46169939d51f40793118 : SUCCESS in 15m 05s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ab98dab4ea02416cbdcad98750b88360 : SUCCESS in 14m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c8d7e559c1c9451d806a6849d04168a0 : SUCCESS in 18m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/29cfba4564dd4c51aa4928d34b1a5a4c : SUCCESS in 7m 09s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3aebe096841849f1a6ec8e50f2ce11aa : SUCCESS in 6m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/874786b32e2e465487da84fb18901491 : SUCCESS in 6m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/34c3b666d6b0415aa035994f51db3f7e : SUCCESS in 12m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f4ae6777aeb44815bd1bb41de591fd2a : SUCCESS in 11m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80a69ed384724a76bfcfa43dd13c5ed3 : SUCCESS in 57m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7f8ead78558045c5a251aec8ef2f981c : SUCCESS in 1h 28m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f01ff9025e0f49018269d97c2d82b57c : SUCCESS in 1h 30m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/ec28ad12f9c34c629190e4d494bc9cc0 : SUCCESS in 1h 20m 00s","accounts_in_message":[],"_revision_number":1},{"id":"5b758989e4152a20877f0902b88d7c900f72f8fd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 16:00:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ffcb1e579159f9f10fe974a3d0968eae41817fd0","revisions":{"ffcb1e579159f9f10fe974a3d0968eae41817fd0":{"kind":"REWORK","_number":1,"created":"2021-07-12 10:12:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/800448/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/800448/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/800448/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/800448/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/800448/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/800448/1"}}},"commit":{"parents":[{"commit":"85ae62b804d28cebb4e4b40822d4464c73523b9b","subject":"Merge \"update constraint for sushy to new release 3.2.2\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/85ae62b804d28cebb4e4b40822d4464c73523b9b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-12 10:12:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-12 10:12:55.000000000","tz":0},"subject":"update constraint for neutron to new release 16.4.0","message":"update constraint for neutron to new release 16.4.0\n\nmeta: version: 16.4.0\nmeta: diff-start: -\nmeta: series: ussuri\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Corey Bryant \u003ccorey.bryant@canonical.com\u003e\nmeta: release:Commit: Corey Bryant \u003ccorey.bryant@canonical.com\u003e\nmeta: release:Change-Id: If6a766c42389a20fd71a768cfc9c718e75265020\nmeta: release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Code-Review+1: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I75f38cab25ad5f82b25eff2c61cbc243a37c8301\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ffcb1e579159f9f10fe974a3d0968eae41817fd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ffcb1e579159f9f10fe974a3d0968eae41817fd0"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
