)]}'
{"id":"openstack%2Frequirements~800449","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~Ibab2c1935df4beef562bba4519f12fee31f72761","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"Ibab2c1935df4beef562bba4519f12fee31f72761","subject":"update constraint for neutron to new release 18.1.0","status":"MERGED","created":"2021-07-12 10:16:30.000000000","updated":"2021-07-12 16:13:31.000000000","submitted":"2021-07-12 16:13:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800449-new-release","meta_rev_id":"c0736eeba93ddd01a4f28aa9c9b04b691fb0d0de","_number":800449,"virtual_id_number":800449,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-12 16:13:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-07-12 14:25:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-07-12 10:23:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-07-12 14:25:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-12 10:23:58.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-07-12 10:32:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-12 11:47:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-12 14:25:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"3a9f933e1032917afb8837b2c50bc541006c1c86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-12 10:16:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"892bf5f2247f2f078e337ca3ad5de0d5b7071c12","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-07-12 10:23:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f56593e7396d07335c74d3e64aabbdb4cfb2b767","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 10:32:04.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/94a369dba7ea4c0daf5c964cd90eaaca : FAILURE in 10m 39s","accounts_in_message":[],"_revision_number":1},{"id":"72f7b0a320c2614c8991a90356bb8e0271972dd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 11:47:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eceadff9489445f8da198e7e635ef3e : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8772d8b01b10436c8cc82091aa2e1170 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7758e358762a496fbacb3901ba1bf6ba : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/013f3603ef1647e5aef0366b04990e6d : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d9f312ab1fec4e9c913376d4d557135b : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e788cfb22732463c95d12a36d4a9a9ce : SUCCESS in 4m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/743bb5331c1c41279e7a63190a6da95f : SUCCESS in 8m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2971e15762ce43739e77dc2cc5f9c8c2 : SUCCESS in 9m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1557f7358c3e4b508fee890d5b87b963 : SUCCESS in 10m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c706ef2c518248eb8e665b22a8908d11 : SUCCESS in 5m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/181bb8be37284cda9ae090a56dff352a : SUCCESS in 10m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4c1643a08c5d460d8281f02949b69bba : SUCCESS in 10m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/64ea681629754906a97bc4cead0297de : SUCCESS in 11m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0fdc72be51d84308b966aac6bd7ff6a3 : SUCCESS in 11m 59s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/f6ad6c0a3f8e47caa3c8cbf5b5b6def4 : SUCCESS in 5m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bf600d5b35824b099dbb71f5bb6ac2ca : SUCCESS in 26m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/44258feb4f3244b49114387298a68077 : SUCCESS in 12m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/da55dfce908e4fc886f583704e68832c : SUCCESS in 4m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/552a5379139b49848f604bf07abddd63 : SUCCESS in 15m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/916be804589f4636a06aabbeea956247 : SUCCESS in 6m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5b66885d42f14d79b59a99ccee577ab3 : SUCCESS in 7m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/129eb713bc8d4f179688d0ea03071436 : SUCCESS in 12m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e0d008512b3249d5bdd7a879eda1e128 : SUCCESS in 45m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bfac09c513144a0abf0af2cd2963410f : SUCCESS in 1h 30m 11s","accounts_in_message":[],"_revision_number":1},{"id":"d3369266f9334d7b7dab315e6370e89ed8e49bcd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-12 14:25:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fe456568fb7b903424973fb6a6c63abfa4d9dca2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 14:25:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9690ed78e380a733bd4d0ebae144a4e857decaa7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 16:13:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/466126af80054041bbd3ae176ff2590a : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6cabef36cd044edba482f1dec531a57 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d06cd9fce17845c78f473a52044b19d7 : SUCCESS in 4m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/69963640813645faa5637bf875ae071e : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/368c5be795e645f5a0839dfc6a868700 : SUCCESS in 5m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e13bf0de3f2f44e6a33498027ae63db1 : SUCCESS in 10m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2f834798a4904e3b93ed2f02bb7ec738 : SUCCESS in 9m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/656d09d5f246452d9a384cd779876eb8 : SUCCESS in 10m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2bfd083a65f14a77836b36f0c2c2301a : SUCCESS in 7m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3ef2913c40a24c758f3958820cb31dcf : SUCCESS in 11m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b04fc2e3a7a849049e1acd046b9f57a2 : SUCCESS in 8m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/aa5824a52e03490eb0e1f809c686d317 : SUCCESS in 11m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d7a534c569d54e7a9fc0f06b9e1ae76d : SUCCESS in 13m 02s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/5b41061320cc4b2e95258598dd582426 : SUCCESS in 5m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ec5408236d6d43fc9d6450a1c1e0867a : SUCCESS in 29m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fa1e1397cdb74d19a823c392ec32b7c3 : SUCCESS in 14m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ad867ef6e6e8428fae136c5f3ef40076 : SUCCESS in 5m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/20c60afbc1574da5b6608ad10345a47f : SUCCESS in 22m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c61f4b5809574e98a9e0f94f50093495 : SUCCESS in 7m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e4c0a985bc59470fb251d6b3c30cdd2f : SUCCESS in 7m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2f97a611f98b483fb25bf3fda04e28dc : SUCCESS in 12m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f42e3a87f38d4ff1a9535ccce1dbbab4 : SUCCESS in 57m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0b5a845c44ae40d0ac9b4084c7b3cf01 : SUCCESS in 1h 41m 26s","accounts_in_message":[],"_revision_number":1},{"id":"c0736eeba93ddd01a4f28aa9c9b04b691fb0d0de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 16:13:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7a930ba1d38fda0be841ade1421641e2612b858a","revisions":{"7a930ba1d38fda0be841ade1421641e2612b858a":{"kind":"REWORK","_number":1,"created":"2021-07-12 10:16:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/49/800449/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/800449/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/800449/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/800449/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/800449/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/800449/1"}}},"commit":{"parents":[{"commit":"951cf48168607d413fdca3c77132ddc8e1b602a9","subject":"remove invalid ansible_python_interpreter variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/951cf48168607d413fdca3c77132ddc8e1b602a9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-12 10:16:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-12 10:16:28.000000000","tz":0},"subject":"update constraint for neutron to new release 18.1.0","message":"update constraint for neutron to new release 18.1.0\n\nmeta: version: 18.1.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Corey Bryant \u003ccorey.bryant@canonical.com\u003e\nmeta: release:Commit: Corey Bryant \u003ccorey.bryant@canonical.com\u003e\nmeta: release:Change-Id: If6a766c42389a20fd71a768cfc9c718e75265020\nmeta: release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Code-Review+1: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ibab2c1935df4beef562bba4519f12fee31f72761\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7a930ba1d38fda0be841ade1421641e2612b858a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7a930ba1d38fda0be841ade1421641e2612b858a"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
