)]}'
{"id":"openstack%2Frequirements~800908","triplet_id":"openstack%2Frequirements~master~Ic3cb82a510d4e6222975f1bb3b64c21cccf0d7d6","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic3cb82a510d4e6222975f1bb3b64c21cccf0d7d6","subject":"update constraint for python-octaviaclient to new release 2.4.0","status":"MERGED","created":"2021-07-15 10:32:37.000000000","updated":"2021-07-15 21:13:10.000000000","submitted":"2021-07-15 21:11:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800908-new-release","meta_rev_id":"17c9ebecea9b7af872e6ec6ecca4d8eda7a63c7a","_number":800908,"virtual_id_number":800908,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-15 21:11:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-07-15 19:30:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-07-15 11:59:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-07-15 19:30:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-15 10:46:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-15 11:42:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-15 11:59:45.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-07-15 19:30:37.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"50eb277a99c956f43d561e5fb1bb04126b480531","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-15 10:32:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3965f16d1adc8424180bad8c5ad874da45cb0f74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 10:46:41.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/07fbb6eea062408986d1986d9e9fd14e : FAILURE in 12m 24s","accounts_in_message":[],"_revision_number":1},{"id":"3e5f911ab9e778fa835fcd5732521103624c573c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 11:42:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6e6adb1d1194728b83edb659317d5eb : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/397b9d6ced494d96b949490b225f8e0f : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b0f869ea0a947dfb4af9e7dec7be604 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00ce1c52c4804c59955dcda543a08eb5 : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0fbcd6fa23e547d4aafc7b2fccb751d9 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4cd84939bc814502a2fd0e7fd6c10aa7 : SUCCESS in 4m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9182de3565754e9297cff738885b7b1d : SUCCESS in 4m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cc16357851384d1fabf7126e2376c591 : SUCCESS in 8m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8277e217021947658c4d3a402113c389 : SUCCESS in 8m 50s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b2bc34b5de8f445d83e67d9508ec7482 : SUCCESS in 6m 22s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e07435694a044ad498b4af7c6dcab613 : SUCCESS in 10m 39s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ddc19e0dfdc4485cb6693c7a6bc0055a : SUCCESS in 6m 44s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/829ede12ca7440b9849b8e55eac1f63b : SUCCESS in 9m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3060cbfa4afb4da39be0c52a8bc17fd5 : SUCCESS in 18m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/03c67906dc4a4c89b7f60fe929aa93af : SUCCESS in 9m 07s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/52da032db5184ddd858cf9c264ece620 : SUCCESS in 15m 34s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a0929c5f92ec444b91953e50231dfbc3 : SUCCESS in 11m 31s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5c60315342f04e378ef4b94a9dc8ba11 : SUCCESS in 4m 59s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/498ef1459a7143b88669c77818967535 : SUCCESS in 20m 44s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/14286eaf3f064e4fa8dc9cd610d664f4 : SUCCESS in 13m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/391431c8700c433dbbf1774e51776cd2 : SUCCESS in 4m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c7afc10087c04c50895121506a4b0dfb : SUCCESS in 19m 29s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/eed2cbff07bd4709869804a002e4146e : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f613b462fd124befa89eac14443ed482 : SUCCESS in 8m 59s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/231c923d1d6e40c49cde5cdf88b698e8 : SUCCESS in 11m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14ddc1931a214df3aa61dda7abb16e5b : SUCCESS in 48m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1172932fde8f46caaa79e124c9f3f722 : SUCCESS in 1h 08m 10s","accounts_in_message":[],"_revision_number":1},{"id":"fffbfe7677e1f4e1e573b76b9de058387af28c4d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-07-15 11:59:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1ab44ca40973c777e4648735f97e25944688b8ab","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-15 19:30:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a593610944dee224ba615e60adbb1d41c980ec01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 19:31:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"10f6fe4fc409f1f42d4bc417fc6f844dc4cf2582","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 21:11:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d1a91599f624852bebc498859d833b3 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/712689970f4e4c54a8ebdf64760242e7 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1390a541c124402691c71f1750f25469 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4deaa17f6b33497c9663730d7ae902d5 : SUCCESS in 4m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/00e011bf0a5d4b80943e9a621aa26c07 : SUCCESS in 5m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/96cb74678e6a4c43896e661ee165c7bb : SUCCESS in 4m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0bde64d0cf0840b7933544f5051efea1 : SUCCESS in 10m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/378847518476446e8dc144fd25cbf4ab : SUCCESS in 9m 14s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f010a90d006c4872821913274a747f6e : SUCCESS in 10m 56s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4a00c5bc7a00406686d2f764fc9b166a : SUCCESS in 7m 04s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3de726c3d9c249179fa4e5a99e28016b : SUCCESS in 12m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cf8af89c64ad4ec6a7840adcb6e1d52e : SUCCESS in 10m 49s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/31512040283d417e9d7ca518c6c6b761 : SUCCESS in 10m 09s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d23a55e76a7048ca97c0822e0c1689f8 : SUCCESS in 15m 29s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/63d1707c28584b35b11f5e4913ae799d : SUCCESS in 10m 49s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/818ce86b41094a408b1568aedc797166 : SUCCESS in 6m 10s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/51325b4edac14a2582f366251057e4f4 : SUCCESS in 26m 00s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/781acab996aa4caa91fec6f7c5941e00 : SUCCESS in 13m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a423801cbc6e46fca854fdf2dcc98589 : SUCCESS in 5m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0eed6f368d1044f6a0f5b434ac6066a9 : SUCCESS in 20m 08s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/48dbc16d48ce44f6b7d816be12bc84d0 : SUCCESS in 7m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4d92bcf47ab9422882c104b129e69b0b : SUCCESS in 9m 41s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/51d55abd413f430fa6102561f85eb7a8 : SUCCESS in 16m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/33a2f674ef164c888cc0d039b9a81f84 : SUCCESS in 47m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/369c88ed1ab047108c67d5ebb464356e : SUCCESS in 1h 32m 46s","accounts_in_message":[],"_revision_number":1},{"id":"19cd11576447eee5953606243420021a89989c7a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 21:11:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"17c9ebecea9b7af872e6ec6ecca4d8eda7a63c7a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 21:13:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2337e36d2d74bf4a2a15f67d021c660 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"33eaed714e2e00525dafc0c7d69318f3fe7e2058","revisions":{"33eaed714e2e00525dafc0c7d69318f3fe7e2058":{"kind":"REWORK","_number":1,"created":"2021-07-15 10:32:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/08/800908/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/08/800908/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/800908/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/800908/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/800908/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/08/800908/1"}}},"commit":{"parents":[{"commit":"5e60818d9f633d950f65d2b47eec2dd10bff80fd","subject":"update constraint for glance_store to new release 2.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5e60818d9f633d950f65d2b47eec2dd10bff80fd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-15 10:32:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-15 10:32:35.000000000","tz":0},"subject":"update constraint for python-octaviaclient to new release 2.4.0","message":"update constraint for python-octaviaclient to new release 2.4.0\n\nmeta: version: 2.4.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I0887820807ede79a0f3c3e7c4bff1b2cc5037c24\nmeta: release:Code-Review+1: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ic3cb82a510d4e6222975f1bb3b64c21cccf0d7d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33eaed714e2e00525dafc0c7d69318f3fe7e2058"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33eaed714e2e00525dafc0c7d69318f3fe7e2058"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
