)]}'
{"id":"openstack%2Frequirements~800962","triplet_id":"openstack%2Frequirements~master~I8de06bbedfcaf12bbdcdb8dc9375f161cfe8e9d4","project":"openstack/requirements","branch":"master","topic":"os-resource-classes","hashtags":[],"change_id":"I8de06bbedfcaf12bbdcdb8dc9375f161cfe8e9d4","subject":"update os-resource-classes to 1.1.0","status":"ABANDONED","created":"2021-07-15 15:32:48.000000000","updated":"2021-07-15 22:06:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9ac812bfc372133de7609d15080af386206180e9","_number":800962,"virtual_id_number":800962,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-07-15 17:20:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-15 15:46:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-15 17:20:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-15 17:23:56.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-07-15 17:24:21.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"d62ef2a8d12bb6dc17f9aab69b888f6aad96f4e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-07-15 15:32:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0460e1bb883bfae711c8fc220e4200ca6a33b327","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 15:46:46.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/488a5cd04d7d4bbdad0266dbd74cb203 : FAILURE in 9m 56s","accounts_in_message":[],"_revision_number":1},{"id":"206ed54f6cf33bbace5d98be81c4811c30fb6995","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 17:20:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f54d439dd0fb47c2bfd483318c38c4f6 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/098e51e137cb488c8c539c48b9ed5646 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbaa10fe03454e5781d06e1708cba6cf : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2f96bd192c347f8ae6997adb465aaa6 : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b0c9d82452443769252e1f2f0432d0b : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1ce7908736b74c568a941378fec65ce1 : SUCCESS in 6m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b432bc9e21df47719b57ed299355775e : SUCCESS in 3m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5697f9da0e1e4fa19b03d43b1db4e784 : SUCCESS in 18m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f598fbbf5a4a4fffa235e2f4ed82223b : SUCCESS in 10m 08s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/8486a9d97f23466aa678d27c19a9ee69 : SUCCESS in 8m 32s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3594bf9234014d0ab48d3a44a3132495 : SUCCESS in 10m 19s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b6d5102c72144ebfae5098a27dd47a2f : SUCCESS in 8m 42s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f548c9de923c4df88c86248392b21d9f : SUCCESS in 12m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/521a9067e11d4c7f9fb24ffb435eb396 : SUCCESS in 8m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2ddd2f15f3a5439ba5bb5f355064dda4 : SUCCESS in 12m 34s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1b6b5e4761a34ffe923f1fea6a7497b5 : SUCCESS in 14m 15s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1b9a5f53b9664e87ac67159243540098 : SUCCESS in 12m 07s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c99ca41444b14650b57526469c790b13 : SUCCESS in 5m 13s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/89039858e2304307b46132b400abb34b : SUCCESS in 28m 11s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c2a59fbd212a4854a72184c7b9c58e06 : SUCCESS in 13m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c84edd01806d48fdaaf067bcf42988dc : SUCCESS in 4m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e18b0ad4b119496b93d205a1be5e25e1 : SUCCESS in 21m 28s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e7c6bf52c60e4454adbd88d81e36c05d : SUCCESS in 7m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/748a44bb5400407a8882bea4fa444e79 : SUCCESS in 7m 15s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/15fef663df0d42d1a847ecc1700c0bd6 : SUCCESS in 14m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7fb2f2004ad14b0fb392f3d86ac5daca : SUCCESS in 59m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/486ab7fd7bf5469c98cc092e97de9b22 : SUCCESS in 1h 40m 54s","accounts_in_message":[],"_revision_number":1},{"id":"9ac812bfc372133de7609d15080af386206180e9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-15 22:06:59.000000000","message":"Abandoned\n\nthe bot version of this patch has been merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"83306c8715bb0b74aee0f6e9c044236d58f2b886","revisions":{"83306c8715bb0b74aee0f6e9c044236d58f2b886":{"kind":"REWORK","_number":1,"created":"2021-07-15 15:32:48.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/62/800962/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/800962/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/800962/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/800962/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/800962/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/800962/1"}}},"commit":{"parents":[{"commit":"5e60818d9f633d950f65d2b47eec2dd10bff80fd","subject":"update constraint for glance_store to new release 2.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5e60818d9f633d950f65d2b47eec2dd10bff80fd"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-07-15 15:32:10.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-07-15 15:32:10.000000000","tz":120},"subject":"update os-resource-classes to 1.1.0","message":"update os-resource-classes to 1.1.0\n\nChange-Id: I8de06bbedfcaf12bbdcdb8dc9375f161cfe8e9d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/83306c8715bb0b74aee0f6e9c044236d58f2b886"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/83306c8715bb0b74aee0f6e9c044236d58f2b886"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
