)]}'
{"id":"openstack%2Frequirements~801069","triplet_id":"openstack%2Frequirements~master~I420930008c831ab277659c0163461fb73a1244d0","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I420930008c831ab277659c0163461fb73a1244d0","subject":"update constraint for metalsmith to new release 1.5.0","status":"MERGED","created":"2021-07-16 14:31:57.000000000","updated":"2021-07-19 05:02:10.000000000","submitted":"2021-07-19 05:00:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"801069-new-release","meta_rev_id":"6dc82a20d21420967b9072c9797d09c6da8ee0b1","_number":801069,"virtual_id_number":801069,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-19 05:00:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-07-19 03:41:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-07-19 03:41:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-16 14:47:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-16 15:53:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-19 03:41:21.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c65ae56606fb243e3c0e7094b501c52e3ee33ea0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-16 14:31:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"602a1f61e20631dc528e918cd63ad29aef17da97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 14:47:52.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/784c5f5f2db748a7bc3c295ea60b26b9 : FAILURE in 11m 42s","accounts_in_message":[],"_revision_number":1},{"id":"90c46b95a5930bfe4b695af1be88996a16187dc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 15:53:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd8ba69ca6954cb590d015b9d4633dd9 : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e05e0a8207c4a7ab09d4afbb67cd8a0 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/545ac5d1cff244e8993eee6cdf9b0163 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66375dacfb7940f3a8428109cebeacae : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b70fa2d398334ebb9b2beb4b1578199b : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/44e9bdedd0c74d6986f008d5a35cca92 : SUCCESS in 5m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/85edfc1b85a147f497edc9c31501f39c : SUCCESS in 4m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3d0034d602d942a3b22006fcd5f8a74b : SUCCESS in 9m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ca04ba05329049cda812cd87eb75e8f6 : SUCCESS in 8m 54s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d1fdfa19e8f44cf0bb768ec502a701ba : SUCCESS in 6m 43s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/54b96e21726b4de1a8ee5de330460a3d : SUCCESS in 9m 16s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3b32fdc850744a919c6b3dbc8c3f70fe : SUCCESS in 6m 27s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8783c1aebb4347629965f6419bcf2dae : SUCCESS in 10m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d445e2283e7c497ab7475ab50b2719c6 : SUCCESS in 9m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8495c131200a45489149646a2755d127 : SUCCESS in 9m 29s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a4a1d13fc5a54aa6929c535522dde03d : SUCCESS in 13m 48s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a956ba12fd104a57810d31efda6f757d : SUCCESS in 8m 47s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f9f4e0d378b642888f8cdd66fa731ecc : SUCCESS in 6m 04s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/cce099391da84a4ea71f418ba3018415 : SUCCESS in 24m 32s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6a93626ae4ea4b308895150b777666e7 : SUCCESS in 13m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5d0d1292afbf4849a7520e0ef9d7f11f : SUCCESS in 3m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dd8cf8cb69f54317bbb85cb50dd1938f : SUCCESS in 20m 58s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/399771bf5e6842988772b01cdd7230f5 : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/30d4b6ef14234c82803ee656cb8b4149 : SUCCESS in 8m 04s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4c80e352a6f84983bfb63512474a40de : SUCCESS in 12m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/319125c4ae68489abb6a3e8a7f2f1415 : SUCCESS in 49m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/886ae461a5044f80a825d53f5194cde6 : SUCCESS in 1h 17m 07s","accounts_in_message":[],"_revision_number":1},{"id":"97279a8fe3c9d9b45e25ccd3e0fe20f69da920bd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-19 03:41:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"db2674b706560eb63adb2448760fe324b49d4ebf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 03:41:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c23d165087c1fa81a7334e147066ba4cb08636c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 05:00:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8afd0f764970463ebacfbb6dab798791 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10772dba39a04b95b9d4fdce992befc0 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/758b94819f2b4877bec1324c4a909b26 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/201bd419323743889dfe86492d884e4f : SUCCESS in 4m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/883e571d88f7462daa944800e893b1c5 : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e241070caf5f4527ad52a458dd436a35 : SUCCESS in 3m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6f464425231f455b9562d0004ae396fd : SUCCESS in 8m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1bb9dc79a85b4c9bb18f4b2da17ce710 : SUCCESS in 8m 06s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7bacdb75258e4b9dad94de846f7fd5aa : SUCCESS in 9m 21s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/570ada4175964d8e8932136d2f3ecf8c : SUCCESS in 5m 11s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f35ab5eeb83346f292e33ab1421f824c : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e06972e368fb45ccb6c5b8722092eb93 : SUCCESS in 7m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6dcdab2f0a5e4dbe9fdbcc0c1cb9b14a : SUCCESS in 8m 25s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/93bbdecb72b148ef876cb22fabef5026 : SUCCESS in 12m 48s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0d81ccd5e23e4f8d98ad065a5502075c : SUCCESS in 11m 07s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0ca98a09280b4fbe805cd5a918a3ed2b : SUCCESS in 5m 05s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9781884084994b9a999b2fd5414bf059 : SUCCESS in 21m 14s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5abdba4c8e0249199b954e77a3842b61 : SUCCESS in 11m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/adf6ef05801b452d93ea8f3ecc8af511 : SUCCESS in 4m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/094ebca74a0143b382d0901e6451ae21 : SUCCESS in 15m 37s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e7125d13e79148b2b528d6636565c592 : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3ae76c1728864a36860278ddda74b158 : SUCCESS in 6m 14s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d2e5fa8204b748edbe6f71968af4e142 : SUCCESS in 14m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb9dc3dbe394430b92760e4f0ff27299 : SUCCESS in 41m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f54788150f34a1684b3e428d34a4b20 : SUCCESS in 1h 18m 10s","accounts_in_message":[],"_revision_number":1},{"id":"959dce0808613a1f9c6a40deb6ef4555567f6a2f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 05:00:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6dc82a20d21420967b9072c9797d09c6da8ee0b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 05:02:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a71f0185272b43c98dfedaacd93c1d14 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e48e749d4e1491401d97717797b1406a80f92bec","revisions":{"e48e749d4e1491401d97717797b1406a80f92bec":{"kind":"REWORK","_number":1,"created":"2021-07-16 14:31:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/69/801069/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/69/801069/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/801069/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/801069/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/801069/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/69/801069/1"}}},"commit":{"parents":[{"commit":"fe3478cc89df9dd9058539069d643ce84af1fdfa","subject":"Merge \"update constraint for castellan to new release 3.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fe3478cc89df9dd9058539069d643ce84af1fdfa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-16 14:31:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-16 14:31:55.000000000","tz":0},"subject":"update constraint for metalsmith to new release 1.5.0","message":"update constraint for metalsmith to new release 1.5.0\n\nmeta: version: 1.5.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I92adc1156365fe49f98df26377f5750a45cf322f\nmeta: release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Code-Review+1: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Code-Review+1: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I420930008c831ab277659c0163461fb73a1244d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e48e749d4e1491401d97717797b1406a80f92bec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e48e749d4e1491401d97717797b1406a80f92bec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
