)]}'
{"id":"openstack%2Frequirements~801071","triplet_id":"openstack%2Frequirements~master~Ia1886b30ec1437d87d5db0435ed391a5df3df5ca","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia1886b30ec1437d87d5db0435ed391a5df3df5ca","subject":"update constraint for python-ironic-inspector-client to new release 4.6.0","status":"MERGED","created":"2021-07-16 14:37:02.000000000","updated":"2021-07-24 02:03:54.000000000","submitted":"2021-07-24 02:02:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"801071-new-release","meta_rev_id":"499d45a9aa8039cc483d73550cb05320f91c0229","_number":801071,"virtual_id_number":801071,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-24 02:02:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-07-23 23:07:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-07-23 23:07:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-16 14:53:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-16 16:08:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-22 13:58:59.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-07-23 23:07:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"9b324b07d6664e5a71958474c4227e9c7b949f61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-16 14:37:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"161620ca5b46616b0f8664fc8a0f869e1c61c3c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 14:53:00.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/23d66053cec741849e3bfe10fe59bc61 : FAILURE in 14m 20s","accounts_in_message":[],"_revision_number":1},{"id":"eb8a0ac2c2d3f36654df9ec0568c3434ffaa4aed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 16:08:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/422c5030bd7c4538a529731af81010f4 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/548bbfa019384474a674d1ce5a1df4be : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a106fc45e724a3285a7853e010b0be7 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3b7409be13e4c80ba4c0f776c44e6ad : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/582dae6be3f64f629cee788909300c72 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d5c719c606134b39b081f320855d9097 : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/443e09cdf3d841e2a8dae86783e1a3ab : SUCCESS in 4m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f1c1b6cb91094c4d8cbb80a82e0b4813 : SUCCESS in 11m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/31beb02018e941c9b8c92eac944388d1 : SUCCESS in 10m 45s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f8f9862137f54b2a846c0fca60aa407c : SUCCESS in 6m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/82590b067e534939b5202c9c43c06b22 : SUCCESS in 10m 59s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/15d3e42dc5984d17a445ba714caf53a6 : SUCCESS in 6m 54s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/cc03daf18e97413bb731baa60d67cbd6 : SUCCESS in 12m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/500b26b1a4a846c5b273483bbebc5902 : SUCCESS in 14m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/18691a7af14441aa850f65f4a7d06945 : SUCCESS in 9m 53s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/08c9206dfe7147b6bc2ecfb312c4bb39 : SUCCESS in 16m 11s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/af1c4d3f0c9740a9854f667859971215 : SUCCESS in 11m 11s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0f336d610d614bb99c34b6006f895a8e : SUCCESS in 6m 27s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/48184dc6ea1740bc977760989dd63d81 : SUCCESS in 26m 49s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/87ec0624769f43f998b163cd92705456 : SUCCESS in 13m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/47ab9a0cc6cd4a5993a340130b2aebab : SUCCESS in 4m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e3b03c2f7e8f40c1b5343ea5774d9817 : SUCCESS in 21m 31s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1ef7ce5f471a4f77839264330e727eeb : SUCCESS in 6m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/20e58eff888f4827b304ed844aacd4b6 : SUCCESS in 8m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8734056b5d60456697a0f39db7ceb620 : SUCCESS in 14m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab6afdc50dee45ada2996dd5d13c8132 : SUCCESS in 52m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4acae9a64cdd42019fe7f52dd5bce1d1 : FAILURE in 1h 29m 53s","accounts_in_message":[],"_revision_number":1},{"id":"d40895ae7ac2afc7e86c2570888aabe36240b425","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-22 13:58:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"46ea5f3915b909414bf262f2eeadb1db42895b31","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 14:25:41.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/eab6655b524649cbad8435b5200e7a37 : FAILURE in 16m 27s","accounts_in_message":[],"_revision_number":1},{"id":"266138f1e715825227cf57007c940de6ce31ff16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 15:26:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8023df95ac1e4f4db7a3859f6fdd0e66 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/496605793c924246ab27441680bed414 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c681f01bbfe4813933b4f25fea4bfa8 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3cfcd6538284bcbac63d254ca20292b : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e7a3cb4312a4cc49b46ec207e3dae54 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/872c541222bd4fa8afb039dc758c2758 : SUCCESS in 6m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/edf88e0e9b064da6b33fca8ef7c7209c : SUCCESS in 4m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/33d29650cb414e6aba18bf64115abfca : SUCCESS in 14m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/23a3ff66713342dda4051456521c1500 : SUCCESS in 8m 49s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/1a631c0ffcf1484ca09eda2a4cdf29ed : SUCCESS in 6m 29s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c54ce6d2cf544586a329d394ceb5cdb6 : SUCCESS in 9m 11s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/6429b9bd6e4149a1adb909471703c918 : SUCCESS in 6m 58s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9cf18720ed7b4e90a20e752324df0dea : SUCCESS in 10m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7bac5d686c264fc2be49e60e9eedb1bd : SUCCESS in 8m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/48d9b3453a054bd2988348ed1ecc42ab : SUCCESS in 9m 47s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/597547487cf04508a3ff3e50414589c2 : SUCCESS in 15m 08s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/df19785260d94b2cb93e87cfb351c23b : SUCCESS in 10m 53s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2a1dca1135e3432c86593a0ac0597afd : SUCCESS in 6m 44s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f8cf7ea973c44cc7bc963239215d1e32 : SUCCESS in 27m 30s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f1c85918e6704f0aa25c0d6398c95857 : SUCCESS in 12m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/63624c16279140f2826e1eb1318988fe : SUCCESS in 4m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5a07144d467b408193984331d77f9e03 : SUCCESS in 17m 02s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/8d9275fa7cd24a34b0c34a0502f959c1 : SUCCESS in 6m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8c47b4e3efda48f8b90ba92412dc9f5e : SUCCESS in 7m 36s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a04997762d644f2a811369a16294cbfb : SUCCESS in 13m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c201288480f4da2a0646828b1e5a94f : FAILURE in 51m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7ff669e2efe0474fb58afb87c006546b : SUCCESS in 1h 21m 23s","accounts_in_message":[],"_revision_number":1},{"id":"be4fa4516898c77f141a006e2b20c239c5dc70d6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-23 23:07:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d642823b89228dfae07df295518ee07cdde6b623","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 23:21:35.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/12fa0f68f64f4e0c8d26e0974c60e2c1 : FAILURE in 10m 08s","accounts_in_message":[],"_revision_number":1},{"id":"ea2ed6896550c1bcdc9ee17099152e681db0d52f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-24 00:37:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad5ffe94dac7436a9027340572117c76 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d891419c976a48a4ad00548498ac084d : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c85682188b04aceb116fc0cba9b1be4 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4313f74ce0bc449f8feb9973c7ab7584 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e4422b68b2544c1989821d5b904cfea : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5705ece6626b4e7a960f78741b920379 : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a48205034f674d8e82082b8c3ff5feab : SUCCESS in 5m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e3fa6bbad9e14441bafa1e0525583a71 : SUCCESS in 9m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3d4711a7c43b40bdbc501992ba6eb938 : SUCCESS in 11m 16s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/acd4155c57d74a68978ae2ede299a215 : SUCCESS in 5m 40s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/629c3c6466cc45dbb03a9ad729c3f24f : SUCCESS in 10m 43s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/fc0054e73d3943b1a65d42b2c9412cd0 : SUCCESS in 5m 47s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5b3eafc0ea2c4751a5a069286599950c : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b7cd8698a69c4cbdbd4e446561b1a258 : SUCCESS in 9m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c7467ceb2ed6467a880a5b454e80f2eb : SUCCESS in 8m 07s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/aa181f10fbe04aceb449069d92d388f2 : SUCCESS in 15m 37s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4b2ee8c3670548019df2042eee19eea9 : SUCCESS in 12m 47s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/006ab2984b0b4e1b9340bd446d9080dd : SUCCESS in 5m 17s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/5e6712be976442e7b405f0731b383401 : SUCCESS in 23m 39s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/da5978b249d541d79b4c9acf6acbc6e7 : SUCCESS in 14m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8ef9e3a9c46e45e78b9d797546904c3c : SUCCESS in 6m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/efdc249c78de4a3d8785ab7f80fbebe2 : SUCCESS in 17m 31s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0cc7601625b1426687b4839d51694ad2 : SUCCESS in 6m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6770bf4fc9484e3bb21612aaa696025e : SUCCESS in 9m 45s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6a03c02af4a84e2a878be73141886b35 : SUCCESS in 12m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c7c88990a904b24969405329fa8a1cc : SUCCESS in 44m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e893848f1cd04dcba6f467e00f19098d : SUCCESS in 1h 24m 20s","accounts_in_message":[],"_revision_number":1},{"id":"362a862c4bf9acc058aaddcd235e0ed54847fe06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-24 00:37:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"89eba32f43af1f54de1ef0501e6aa7d663c04256","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-24 02:02:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c44d2994f0104770a84e0f0ed98673e6 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af3f6b13df78465ea07058ee7f56cff5 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d2c671c40c14d8f88b43e0647f2af33 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9645e7b9f9ea4e36b6f9f5f511548358 : SUCCESS in 5m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/caece86618244f1c8a70e0ccb30ac7e2 : SUCCESS in 5m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1e6d2291c4bd45d3ba8fe72d41d49a0c : SUCCESS in 4m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/696edade88044627b2fb5d7817338e5b : SUCCESS in 10m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/271b897881de4c0c8d653aff9906191b : SUCCESS in 9m 35s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a3bd4692928c4df6bea47d684fa2536a : SUCCESS in 9m 19s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b24e248e21784ad484a711fdbd0a2ee2 : SUCCESS in 6m 14s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/76f89832fe3a4e5cbccdc0c9cc5ea498 : SUCCESS in 10m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/83c19a1d3f3444cabe4546b6fac86049 : SUCCESS in 8m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/50b15b93d29b443eb29c31e681f80aed : SUCCESS in 8m 45s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8224450167c14afe94ef3600622c73d7 : SUCCESS in 13m 38s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7e67e84cd8b14e6e88e78aacb612163d : SUCCESS in 11m 47s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/acadffd577ab4479856b0dfa7c79a3f1 : SUCCESS in 5m 58s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/564d9914f56e42a2b1fb384889dc64e3 : SUCCESS in 21m 35s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a65e9695e5a04ca7a6795c9a9f8cfa47 : SUCCESS in 11m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/04f65dcf5b1342a3b4a1365a7585b787 : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6655dd9ec9a8495e8f453396f3c388ee : SUCCESS in 15m 55s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/dfa4027f5090412da509c4ec991842ad : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/666725ad26e348d382a80836248eaf27 : SUCCESS in 7m 24s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/03ae28c89370424fb421ec6fd5796c42 : SUCCESS in 18m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54b33a998322497c9bdc4228b522f329 : SUCCESS in 42m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3ed14d0b4a0b4b7fbbba8a27936343fb : SUCCESS in 1h 19m 26s","accounts_in_message":[],"_revision_number":1},{"id":"727533525e05111637463843e082da84d0c5a6b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-24 02:02:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"499d45a9aa8039cc483d73550cb05320f91c0229","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-24 02:03:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22e737d00fe04556a54e09c92b753e34 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4911cd09d42ca0f257c4a84dd35982073c5df6f2","revisions":{"4911cd09d42ca0f257c4a84dd35982073c5df6f2":{"kind":"REWORK","_number":1,"created":"2021-07-16 14:37:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/71/801071/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/801071/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/801071/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/801071/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/801071/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/801071/1"}}},"commit":{"parents":[{"commit":"fe3478cc89df9dd9058539069d643ce84af1fdfa","subject":"Merge \"update constraint for castellan to new release 3.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fe3478cc89df9dd9058539069d643ce84af1fdfa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-16 14:37:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-16 14:37:00.000000000","tz":0},"subject":"update constraint for python-ironic-inspector-client to new release 4.6.0","message":"update constraint for python-ironic-inspector-client to new release 4.6.0\n\nmeta: version: 4.6.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I362996aedfd0f58e234df28b8d4bbde695904d56\nmeta: release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Code-Review+1: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Code-Review+1: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ia1886b30ec1437d87d5db0435ed391a5df3df5ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4911cd09d42ca0f257c4a84dd35982073c5df6f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4911cd09d42ca0f257c4a84dd35982073c5df6f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
