)]}'
{"id":"openstack%2Frequirements~801073","triplet_id":"openstack%2Frequirements~master~I0a7c9f46ec0cce7cd4e446e9905ef4743cd3287e","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I0a7c9f46ec0cce7cd4e446e9905ef4743cd3287e","subject":"update constraint for openstacksdk to new release 0.58.0","status":"MERGED","created":"2021-07-16 14:41:25.000000000","updated":"2021-07-23 07:34:07.000000000","submitted":"2021-07-23 07:32:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"801073-new-release","meta_rev_id":"bab5416aa6ceb764a1d51e0757137a56b48b8816","_number":801073,"virtual_id_number":801073,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-23 07:32:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-07-22 13:58:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-07-22 13:58:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-16 14:55:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-16 15:55:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-22 13:58:51.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"57069bd620cfbe8e3613a18ab397963a41362cd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-16 14:41:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7406cba4416c7cd607c91bc89fdfba27359f7992","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 14:55:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/436f17e5b90e468fa5cb5b7d1e937a9a : FAILURE in 10m 11s","accounts_in_message":[],"_revision_number":1},{"id":"007126a4f988a72e9830809d6cd6dc14cd8d8e5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 15:55:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b744aa3317fe49d384a3925ca5dbbedf : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5953b38ac67e4dc987925cfb9b268d92 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1179ea31ec054195a153b3054b1559b9 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7cc73fb741404983a94291b8f4edc268 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a58990d574849e8bd2729dc07e0282f : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a4d166d6b6a48ffb2d51310b54fca2a : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d9fcdd3318b84e7598cbf97d0703fd32 : SUCCESS in 6m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7736ae82c31d4324866320f7bc79ed6e : SUCCESS in 9m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/efa2c180daba40319e17f88366f1b493 : SUCCESS in 10m 29s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/69cabefa9892492d80aec28506d40e4e : SUCCESS in 6m 15s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/566237964a9d4b7baa91cdb8fc9f349f : SUCCESS in 8m 58s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3511b1285652475aa3b928d0209e0ede : SUCCESS in 8m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/78b86415819e43f99a408981c6fd5b7e : SUCCESS in 10m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/aa64a09301ba4e0eafd4ef65b393146b : SUCCESS in 18m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a5b746d2cf81458baaec8d7493771291 : SUCCESS in 8m 51s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7e42021cb83e476894f248176089c7e5 : SUCCESS in 20m 21s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/849f82af24eb4770bf0e26068af076ef : SUCCESS in 11m 15s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0f7e6be09ea944c18495fa2bc5997d23 : SUCCESS in 6m 04s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/275ea5d1a9484684b0e53bf1f4c20453 : SUCCESS in 21m 40s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/93568c7e823c4e70825106bc7d3cb089 : SUCCESS in 13m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c0837c07ceb54782b9c99ed0c2732682 : SUCCESS in 4m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/76a8a82d65794aaba051526833234c56 : SUCCESS in 19m 07s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/05544cdb066044ebafd04a5b591d80bb : SUCCESS in 6m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/008b86f0ca27430b837aa10a333a347a : SUCCESS in 7m 02s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/90a1da8e02494ec8aec4b9d97a03dcdc : SUCCESS in 14m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa38dc72de1444eaa99a121e406c82fa : SUCCESS in 47m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/201c70e762bb42809f66e6dcdc16f2f3 : SUCCESS in 1h 12m 23s","accounts_in_message":[],"_revision_number":1},{"id":"8bb5b58fbcc27a920f27594edefdbdec2c029dad","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-22 13:58:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f5c48cd7b38cf63b0eeb2d45fbe3953f9b967a75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 13:59:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a4e6b9d4c64bff3028aa71ad831b71d4a3c180e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 15:36:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38d384e16d0c44ba8f296344aa0353d8 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f7f9662a4c1495e97f9dbd6cfcf42ef : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f8c3bd1b50949e6ac1df06d5c137c46 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7fd793dffa14789943b978e2d161d7d : SUCCESS in 5m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aebf746357294b91be35560b408ba251 : SUCCESS in 5m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4e0bc81455954f0aace1761d1350c00c : SUCCESS in 4m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9c7764606fb54357b49739f0d31874eb : SUCCESS in 8m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/487b5a1810ad43e39cec19dc8661793e : SUCCESS in 9m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8430db467a8046e4b33e468118b9010f : SUCCESS in 9m 06s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7097cf04e1244e77bdb093c4e8b88806 : SUCCESS in 6m 22s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a6c73090b6ca48b5ab3c46b7700b0db7 : SUCCESS in 10m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c39664c922cf462d8bfb5a2dc555a001 : SUCCESS in 9m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c9ff04d9e8ad4302933e9335803323b1 : SUCCESS in 9m 28s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b1c55267d8fe49ea8f59aa5f42ab1e49 : SUCCESS in 14m 19s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/37f5ca59f6214f879f5e32af8689f751 : SUCCESS in 9m 16s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/810a2659132243da92d17259725463d8 : SUCCESS in 5m 37s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c74f477a8fbd45499a043a10cfca3b9c : SUCCESS in 23m 59s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6ac0b2f49985491d90af6c5fced3d815 : SUCCESS in 12m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bd7dd7306f404d36af62ea3f20aadd69 : SUCCESS in 4m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e44fdd3ccbe345b98eba023a9978efd8 : SUCCESS in 18m 04s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/62c9e01a1e7a40b6b378fb41f8ea4cec : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/81718cd11b674470b9f9be4cafe834c8 : SUCCESS in 6m 57s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cf611f279c9f49b38d6eb23be6d9f8ea : SUCCESS in 12m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85f2c2ae137247c783db5d429a491e71 : FAILURE in 44m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a17528b034594ec1b6160668dbf97667 : FAILURE in 59m 21s","accounts_in_message":[],"_revision_number":1},{"id":"39a75117f54274f0b562ae77acac3d83d1b7746b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-23 04:29:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f2775d76bf3363bbb2af5f2a3547c11d2e6c90cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 04:49:17.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/be67c115b86e4d3ca378368ee13bc481 : FAILURE in 9m 23s","accounts_in_message":[],"_revision_number":1},{"id":"b02d80f4da0085d660f54df90048de346dd3b76a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 06:06:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68292a202c2046d5ae8222fcb06b03f0 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88a414d68e7943f3afaf88f9c847b9e5 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1b40f5546254b229b93d4884bdcca5f : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37a6dce32e364044a803dfc523f300b3 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6948c572e71a4f89b3c2bdf29f50f101 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c698d3263d5448d7b599fde2d55d42bf : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6937d9618a0e4003b7a57151bff7a3de : SUCCESS in 3m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/24e973348f91491db6b305b943065c98 : SUCCESS in 10m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/eda20ec11eda469281accaa8f007e91c : SUCCESS in 8m 58s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/54615cdab50b4965a53fd72e8dfbccfa : SUCCESS in 6m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/68407adb796b46d6b8dc51689d3eb60b : SUCCESS in 9m 02s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5980b57df84a40e28c3c02d85394a03d : SUCCESS in 5m 45s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8c7e66c7bd3e4c02ab533144ae8fd426 : SUCCESS in 11m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9b10691226a644728586470b746c0888 : SUCCESS in 7m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4a065c548a954703a170da98344a4f80 : SUCCESS in 8m 34s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/67b642e3c8eb4e6e908e6d6043006674 : SUCCESS in 13m 35s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/310d2f4168ed4a9181e415d08a8e4ab0 : SUCCESS in 10m 47s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/fe0181e6d1a84fca8cbf84b25d030916 : SUCCESS in 5m 06s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/097ede764c184d49a4f2d366e99dabaf : SUCCESS in 25m 02s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/20780c68b8d345e8b35c32aca15211d0 : SUCCESS in 11m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/91c4935ae69f4f87a44ebbe6a4e0c3ef : SUCCESS in 4m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ea95be9914944347b988382700b5cc62 : SUCCESS in 19m 52s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f27703003199416b9b9aef0d826e275c : SUCCESS in 7m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/52b94a35ac304a82b52deef8d330b08a : SUCCESS in 7m 06s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b31c2066b1bf4c79911b7b093673c646 : SUCCESS in 12m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0af827487fd2463ebfca99fe3df0660a : SUCCESS in 46m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff24b8cdcda34ba993359e4d1c6677e5 : SUCCESS in 1h 29m 56s","accounts_in_message":[],"_revision_number":1},{"id":"7ccfcaf8ade8cdbe0851127e67ecafdb7259829a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 06:06:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"15ae283940fa602e2ee90f5524cf69dff7aa802e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 07:32:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f3239fd9a254190864e13359b8d2eef : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d2eeb74e2994359807b1929a37de453 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebe984301b2f4b6292d19c941d87bcb5 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a66680acc6549e69dcd45f0698bcb39 : SUCCESS in 5m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5c0aebbe1e6f4e959bfc9bde22fefc33 : SUCCESS in 5m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1f00811024934918aff8280dfb2cec88 : SUCCESS in 7m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e31d886bf9e94672bb3224d93abf362a : SUCCESS in 10m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/908f8df1b6a54ac5ac5838237b2f2bd4 : SUCCESS in 12m 46s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9263b8dc402b461088b2e3d1a41647f3 : SUCCESS in 11m 52s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a82441b655f049eb842bddb08f7b3c7e : SUCCESS in 9m 49s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4d5df2aa49044ff785b892934ffe564f : SUCCESS in 12m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/46bfd23acf9f4a38a3c83a5e88ece466 : SUCCESS in 11m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/97183249818c48cd9b69d53f8b878329 : SUCCESS in 10m 17s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8af9cfee72004ba7837fc3f2e8f1de7f : SUCCESS in 16m 19s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/29e9b217fa574a7b843c88b01a5060c4 : SUCCESS in 11m 32s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/cd2c41dfbfe34c279e33c1660ab0dd0a : SUCCESS in 6m 51s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/767b0666c26c4fe581b25df3f5141725 : SUCCESS in 29m 35s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9717bea88de94ab3a4451169a89ab40b : SUCCESS in 13m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/22ba26458e02416b97add73ba55727ea : SUCCESS in 6m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/537b634e0e03416083b01d517fa43e21 : SUCCESS in 19m 35s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/ddc2e7be0d8549e29ab3eec3b502e6bf : SUCCESS in 6m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/97a3bde0ba034d08a6b15296c0d98878 : SUCCESS in 10m 44s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/293763676e2340e8abc8527c78a9c2ae : SUCCESS in 13m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c90796e4ae96493a81d0e31267d10baa : SUCCESS in 47m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c4894cc1b4704b02b5b98dd85d81b86e : SUCCESS in 1h 24m 33s","accounts_in_message":[],"_revision_number":1},{"id":"902fa5cc0cf751deb6fc8cc15b1c7082377a2519","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 07:32:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bab5416aa6ceb764a1d51e0757137a56b48b8816","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 07:34:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f968d9534e245f897e5b4399d76aa93 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"05c94df5858139270462689dafb9aec5fccfd433","revisions":{"05c94df5858139270462689dafb9aec5fccfd433":{"kind":"REWORK","_number":1,"created":"2021-07-16 14:41:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/801073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/801073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/801073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/801073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/801073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/801073/1"}}},"commit":{"parents":[{"commit":"fe3478cc89df9dd9058539069d643ce84af1fdfa","subject":"Merge \"update constraint for castellan to new release 3.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fe3478cc89df9dd9058539069d643ce84af1fdfa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-16 14:41:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-16 14:41:22.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 0.58.0","message":"update constraint for openstacksdk to new release 0.58.0\n\nmeta: version: 0.58.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I290e9aa7a1e9a42fc4c7c2f2466366e4c5a51fa7\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I0a7c9f46ec0cce7cd4e446e9905ef4743cd3287e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05c94df5858139270462689dafb9aec5fccfd433"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05c94df5858139270462689dafb9aec5fccfd433"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
