)]}'
{"id":"openstack%2Frequirements~801273","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I06fbdd851fad2e330349d646eeefbc08b63f3a9c","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"I06fbdd851fad2e330349d646eeefbc08b63f3a9c","subject":"update constraint for tripleo-common to new release 15.3.0","status":"MERGED","created":"2021-07-19 09:46:40.000000000","updated":"2021-07-20 00:30:42.000000000","submitted":"2021-07-20 00:30:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"801273-new-release","meta_rev_id":"abf69244a25a82d7518250fbcbab9d8172078d67","_number":801273,"virtual_id_number":801273,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-20 00:30:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-07-19 22:55:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-07-19 22:55:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-19 10:05:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-19 11:13:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-19 22:55:56.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"96de761210d5492fcd03ff7b1bfbfc9feeef48e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-19 09:46:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d018ea3a84f8c3ce83c5c782fd650492fb0c524","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 10:05:03.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/730695a1f8af4b0abf326e77c1023d1c : FAILURE in 16m 40s","accounts_in_message":[],"_revision_number":1},{"id":"52fefcb9dfdc682381ebb9b991d483a95cdc5fac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 11:13:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aad7a66f4b045dbb518258b5ecd50d0 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ee6dc5a10dc467bba5e9d5b1f99ed4a : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/996290da4fd441228c95e3a3cad8968a : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/763ac030a8ce4f84bdd91f366281dcc5 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/513e2930e79f42f3bcf865497c615189 : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ccd42830770f457da1ef274c72f9e1e7 : SUCCESS in 3m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/393ce53d4d244f40885f9cbffdf733bd : SUCCESS in 11m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a229de3fcc70467898b9e8e59bb84e5b : SUCCESS in 8m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e17ece6a745447c094e4b10b0b24b583 : SUCCESS in 10m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/40a2694c39064f339f95123223a8d276 : SUCCESS in 6m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c3837ccf129c4966ab24e21ee8bee6dc : SUCCESS in 11m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d095e6797f7a4e9cb9c3eeb90c75c47c : SUCCESS in 7m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d610e6a72f444a219c8267eb540bd45e : SUCCESS in 10m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f13ba4e672084510a67dbd7f208934d1 : SUCCESS in 18m 42s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/f55dd00153654658b96095978c6e9999 : SUCCESS in 5m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2bd9f10ef7a442a5bfec12be919068da : SUCCESS in 22m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/61f3e92ff7ec42bb8cbac73b1da60c8b : SUCCESS in 13m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4b5bd7375779470f8ca4b640501c9a81 : SUCCESS in 4m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d7ae21197de54947905315033b86bd6b : SUCCESS in 17m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/02efbaef68c9445a8680f4bfffc50d31 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/42c044ff8f904e4ebc20c5102ea5a894 : SUCCESS in 6m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/506f0f33450f46c5bd914c48d89d2bad : SUCCESS in 11m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d42bda75c014e00abee937ef359e897 : SUCCESS in 46m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a2e893f3c204917b2865b33f75b16a8 : SUCCESS in 1h 26m 03s","accounts_in_message":[],"_revision_number":1},{"id":"e1b5617005b52f47d9f28afd8677861edc487cf5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-19 22:55:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bd0e4885edeea9c58127a7911c515794eea4ab84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 22:56:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dd638917b899dd844ead946533a4d8e590dc8624","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 00:30:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51426754ee02423281357dd7a228b338 : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13d3d13e4b1045149df8904f37df3013 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2dd9815008e43719a5f1617ca59e40e : SUCCESS in 5m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b422066420634283b94adc6eb03cbcba : SUCCESS in 5m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5cc0577170d246878e5dca4a2be9ec2c : SUCCESS in 5m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/793fbf7b77214d1a967f08611db6be70 : SUCCESS in 11m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/38442d759b594f7dac167741b91f1bc2 : SUCCESS in 11m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a1082e75fedd4f5f8742aa2127fc7cfd : SUCCESS in 9m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e4e6514c19154e89985e6e475167c288 : SUCCESS in 6m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/469a554cf1da47cdb50e9bcfa2018b24 : SUCCESS in 11m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c1afa3d59bb44fd580c71419a6a3dec1 : SUCCESS in 8m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2715b5316fd143b2ac62e52321f5f73a : SUCCESS in 11m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/eaa46f9f28624559b692c3b0b539a899 : SUCCESS in 14m 32s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/b9e416f1b12a4e33ad749c975ffc8632 : SUCCESS in 5m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7548de36b72c4efcb3f120a9eff33818 : SUCCESS in 31m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f53de537cbcb417a9d6c868fedfd6db2 : SUCCESS in 13m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/07fdb81ffc284a9a8ab334a0316ddb9d : SUCCESS in 5m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d3b335f038b44b41a479a392cf536f08 : SUCCESS in 16m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3d9e7457635f4c8ea3f2de1d18cbc6a9 : SUCCESS in 5m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6bccc3943735451ca56d95a289674f53 : SUCCESS in 8m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5f7e931dae0f4336835c754b719d79b5 : SUCCESS in 12m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bed4cac0df2f41c6a5e5b3f83412c4d2 : SUCCESS in 55m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4a7672ff6bab4642a04408bd2c86d49d : SUCCESS in 1h 32m 29s","accounts_in_message":[],"_revision_number":1},{"id":"abf69244a25a82d7518250fbcbab9d8172078d67","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 00:30:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f544bf7a70721f602d3ad30890e058f09b9e00ea","revisions":{"f544bf7a70721f602d3ad30890e058f09b9e00ea":{"kind":"REWORK","_number":1,"created":"2021-07-19 09:46:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/801273/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/801273/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/801273/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/801273/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/801273/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/801273/1"}}},"commit":{"parents":[{"commit":"dab8ba4a0fc3d4ad57c51757d93df8916e13a489","subject":"Merge \"update constraint for python-octaviaclient to new release 2.3.1\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dab8ba4a0fc3d4ad57c51757d93df8916e13a489"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-19 09:46:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-19 09:46:39.000000000","tz":0},"subject":"update constraint for tripleo-common to new release 15.3.0","message":"update constraint for tripleo-common to new release 15.3.0\n\nmeta: version: 15.3.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: Iba00fee26b60049f34e9c9dac43433b507de1404\nmeta: release:Code-Review+1: wes hayutin \u003cwhayutin@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I06fbdd851fad2e330349d646eeefbc08b63f3a9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f544bf7a70721f602d3ad30890e058f09b9e00ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f544bf7a70721f602d3ad30890e058f09b9e00ea"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
