)]}'
{"id":"openstack%2Frequirements~801284","triplet_id":"openstack%2Frequirements~master~I7798f19f7cf1b16ad905d9403c9b5fe51abea73b","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7798f19f7cf1b16ad905d9403c9b5fe51abea73b","subject":"update constraint for sushy to new release 3.10.0","status":"MERGED","created":"2021-07-19 10:53:28.000000000","updated":"2021-07-20 00:17:03.000000000","submitted":"2021-07-20 00:15:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"801284-new-release","meta_rev_id":"876d05c1f5489ff4fc60c5543d646b2e6212fd8b","_number":801284,"virtual_id_number":801284,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-20 00:15:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-07-19 21:52:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-07-19 22:55:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2021-07-19 12:50:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-19 12:07:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-07-19 22:55:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-19 11:07:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-19 12:07:20.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2021-07-19 12:33:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-19 12:50:50.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-07-19 13:48:59.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-07-19 16:57:17.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-07-19 16:57:17.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-07-19 16:57:17.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"707a16a1ef43da91a1be991fce68f535d6186583","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-19 10:53:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46e4e9549a515afec612037b11a9b0a7f87836dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 11:07:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4347aab14a7a44ae9c534f43450a0698 : FAILURE in 9m 35s","accounts_in_message":[],"_revision_number":1},{"id":"4923702a94ebf84a8e4111dbb96a534a9cf0bd6c","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-07-19 12:07:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9f54c2bb781dee1fbc7cd1d6683ab4136f138623","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 12:33:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8486f8b251544511904d19c33d6d1a99 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/078f8b57f2564e4c998a34ab97019594 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77134a9441614ad098fbeb88c615ae4c : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97cf0de10df24b8792b8910f17a34346 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd18afe56c954eeda76467f0eafba34b : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/44e1b8f59f674f1fb1c13ae61c71db21 : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cd76c222fbb34c3090b8386999300429 : SUCCESS in 3m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/74428744f53447c082dda1313a9c89f6 : SUCCESS in 10m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/eb95fae0d2e74d6094683169417cc2c3 : SUCCESS in 9m 36s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/642b580fd6484941b12f284178926eec : SUCCESS in 6m 00s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/89de51e859d342619ef6971d90c244a2 : SUCCESS in 9m 10s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2080e7c8af064aed9c6c5d4a5c1963bc : SUCCESS in 6m 20s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ea92e507c0e94284a86bf903ac90b94c : SUCCESS in 10m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2444675ea8104a00ba3545be57a129d7 : SUCCESS in 8m 23s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d74cbfed0fff49168057f0cdb2af9c93 : SUCCESS in 9m 43s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/732c3a5ffbd24b7ea194b07768b5fe7a : SUCCESS in 15m 06s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/3fe1c8c1ce9347bca8c1d263e7342ac8 : SUCCESS in 8m 46s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b7deceacfb984b45ae833078f2876ed9 : SUCCESS in 5m 13s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/66dff5f55ac94f19b76ae95d2e17a7dc : SUCCESS in 24m 50s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d3e8db21437643b3b331bdeef2aa16a1 : SUCCESS in 12m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/483dbfb496684914b366036e85a5c293 : SUCCESS in 4m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/abc5abe15304414ca5e106f672b9aabf : SUCCESS in 18m 44s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c61e197b91c14173bf31909b5d934d3d : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5221ec20074c4f36866d84f27d1575cc : SUCCESS in 6m 40s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/33aef06dc3c9404690e86a2e1af1ba7f : SUCCESS in 12m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/185db3705c83418182fbfdbc0cc58c10 : SUCCESS in 49m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae42b60e237a400da6efa726737815bf : SUCCESS in 1h 34m 00s","accounts_in_message":[],"_revision_number":1},{"id":"396968242f59766264146e6b55826a540bd5f641","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-07-19 12:50:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"723b9317e9bc0bec864bd6ed1d1673b936cc48bf","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-07-19 16:57:47.000000000","message":"Patch Set 1:\n\nHello everyone, we need this because we will cut the new bugfix branch for ironic, tks!","accounts_in_message":[],"_revision_number":1},{"id":"95f80b1abaa85dce254eee577e0770a7b28f0f39","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-07-19 21:52:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"848d85fbc38d6b71d59a5b2de1fb89a4713afac7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-19 22:55:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"27c43075c107bc099eb5ced946ca7b133da67b7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 22:56:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"566923a81154b1e989581be8f3556b83b0742336","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 00:15:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75c6586b4ea1442584c6b308a1e0fae1 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1b36ff1208d4f089962ba3777c97bb0 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9894707c78a645b2930923f3b189b6ea : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c511191908b4d6c8f05af854f1b969e : SUCCESS in 5m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3a2119ea024f4840a9ee151af2c47427 : SUCCESS in 6m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4b1cfc05ae5d48c8b8601ab2f1dc879f : SUCCESS in 5m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/acf4a18233b040a3bfc5a1838b7cf382 : SUCCESS in 11m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6605a5765b024c27bbba38e0843047b7 : SUCCESS in 10m 40s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3f4d508443ca45c08bfa30f8b9173bce : SUCCESS in 11m 01s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5f9f4133580b4da585a663590b1ace85 : SUCCESS in 8m 00s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/97ce77cc788648cbab20e968ef2231af : SUCCESS in 10m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6e8c214bbdda4d9d9437f7d6170d931e : SUCCESS in 10m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e06a9526a1b64f62b5f9f2e63fff4b43 : SUCCESS in 10m 08s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b2abc7dbfbbb441e8a380346d561c007 : SUCCESS in 16m 09s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/56788b5503994960b09e576cf470b3a9 : SUCCESS in 11m 21s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/231b14cd663649319436b01bb97cd449 : SUCCESS in 6m 01s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d7eea32252d94036b9d5d5887dbacc8e : SUCCESS in 20m 30s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/4727326749c6439fb56c1414e1ef2a2a : SUCCESS in 12m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/eaea311368d84061a1eb7b6480cf20dc : SUCCESS in 5m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bd5fe98a97404985bd2d0f1327217f7b : SUCCESS in 17m 20s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f8d77453013145eab8429df14a1e0440 : SUCCESS in 7m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/932398ee4c6a4d9c9487e66dc1129132 : SUCCESS in 8m 21s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d32b3df095db42928b17b4844303117d : SUCCESS in 14m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45d7b9f987724b778df627e003626e8f : SUCCESS in 44m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f48712043c984af586a00fcbdec312e1 : SUCCESS in 1h 17m 37s","accounts_in_message":[],"_revision_number":1},{"id":"6eb4b35d3c5b802799375f8acb7a9f07a5c82ccc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 00:15:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"876d05c1f5489ff4fc60c5543d646b2e6212fd8b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 00:17:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1c130538ff5443481d4662400cbde67 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"abbd4b407d8fd3ce39659de7d193887395204cae","revisions":{"abbd4b407d8fd3ce39659de7d193887395204cae":{"kind":"REWORK","_number":1,"created":"2021-07-19 10:53:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/801284/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/801284/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/801284/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/801284/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/801284/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/801284/1"}}},"commit":{"parents":[{"commit":"f398f5874bf895e7aca4759bedfb214947ec8a90","subject":"Merge \"update constraint for metalsmith to new release 1.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f398f5874bf895e7aca4759bedfb214947ec8a90"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-19 10:53:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-19 10:53:26.000000000","tz":0},"subject":"update constraint for sushy to new release 3.10.0","message":"update constraint for sushy to new release 3.10.0\n\nmeta: version: 3.10.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Commit: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Change-Id: I77ea6ed3bb68aab663e93fa3a5adb4246ae84aa4\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I7798f19f7cf1b16ad905d9403c9b5fe51abea73b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/abbd4b407d8fd3ce39659de7d193887395204cae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/abbd4b407d8fd3ce39659de7d193887395204cae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
