)]}'
{"id":"openstack%2Frequirements~801768","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~Icbc085ca48de9d53c10c502c3980d4dfe0db0e7e","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"Icbc085ca48de9d53c10c502c3980d4dfe0db0e7e","subject":"update constraint for os-net-config to new release 13.2.1","status":"MERGED","created":"2021-07-22 10:53:28.000000000","updated":"2021-07-22 15:36:34.000000000","submitted":"2021-07-22 15:36:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"801768-new-release","meta_rev_id":"3c1b1e08f99f34c975007dff27686407ea64ae24","_number":801768,"virtual_id_number":801768,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-22 15:36:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-07-22 13:58:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-07-22 13:58:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-22 11:15:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-22 12:27:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-22 13:58:50.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c7692242b1de091ca9f605b9713beb5e294a05ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-22 10:53:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50e860d7dd25c50c85c4933caa979624c1310022","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 11:15:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b799dd5ce05e469686354c4e7c145563 : SUCCESS in 17m 51s","accounts_in_message":[],"_revision_number":1},{"id":"c95f1ab4c06d83c2351b3f72bbdf8586d299ceb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 12:27:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f008727038ac4eb7940ed0974a507bfe : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36ee4347596e4705bfeac774d017b8a1 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/624f9b3418f249d3a716a6e40c586cce : SUCCESS in 5m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fad7eb6502bc46a1b7e54bee3c6aebd0 : SUCCESS in 5m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d1e926f9726f46a6b02748406a53ba6b : SUCCESS in 5m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1e9456b3cb2c4d12ae0be1a26fca04b4 : SUCCESS in 10m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c8b19bc20b1d44b2a02cd48d088e89f8 : SUCCESS in 10m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b1b264ae4cc7402692bfac2e1e231277 : SUCCESS in 9m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/72da20219eae440695bc82dbfc22c9e6 : SUCCESS in 6m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c214cc4f5afe4d1d8effd379d80a200e : SUCCESS in 10m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0b66894444d646e08aab867839802c8c : SUCCESS in 10m 02s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/06e0b653305245c5bf49fc2e7408314a : SUCCESS in 8m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3b5f102c71464093b76c43c3b0798f32 : SUCCESS in 17m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5e1f50f89dd34fbfbb1d161dbd33abe8 : SUCCESS in 23m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9a3320c89a924bda87e876e48b9f90c8 : SUCCESS in 14m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4d87fae38f744c4ea3272ccf13a3c0ae : SUCCESS in 4m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3df1569986a74882826aee6c6b9d40dd : SUCCESS in 16m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8d474c219f2f493fa60b368e8f9b2e78 : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e228c09fb2f64d58aac29f1093366f96 : SUCCESS in 8m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9ee4f8b57dec41b797e569987d22cb6a : SUCCESS in 12m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f03a1bebe67b4d7ab53b3a815ba831b1 : SUCCESS in 56m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf8fa00203764222958a2793eab13266 : SUCCESS in 1h 28m 48s","accounts_in_message":[],"_revision_number":1},{"id":"4fdd49979b3fae0ba6a216608e8a2d83454bd938","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-22 13:58:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7abb1d93e351906c2964a56870e5aa2922c965ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 13:59:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e366f6fd86b417168e58b6c1938e0a28ccb47de5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 15:36:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5d27e642d4f4afc88b36ba188d19b08 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e38ee9c36e0d46ada72695d09ae348fa : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88e25cfefaca49f5bdaf994c902c3498 : SUCCESS in 6m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b343707ab942441dba4b2a5da172a29c : SUCCESS in 5m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ab6625bfff354d16b5af9a4e092347bf : SUCCESS in 4m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8bb4cc4ec2284202b9971c7802e485de : SUCCESS in 9m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a46dbb5f27ca40baa9eac37203f9ecad : SUCCESS in 9m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d0c92b7240ab482195c9bbd437e94d13 : SUCCESS in 10m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9fd4aff5ec1742bfae924d1cf2d25b49 : SUCCESS in 6m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d34b4f83f5d048d4a53b09036c3ce2b9 : SUCCESS in 12m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/18740c90613743e49cca280b78667315 : SUCCESS in 11m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b6cd19e342854eed8097611f00dfcdc3 : SUCCESS in 9m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fd57067d4bbb463c8c469874b472c348 : SUCCESS in 13m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ce7a4848862d496a92e80c291b03d078 : SUCCESS in 22m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/70a6042847ff4b7794e09541f5d6f8c9 : SUCCESS in 12m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ab698c771aca4477b1cfa5e7871e1063 : SUCCESS in 4m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7b9eb034b885489ba68a05dbabd7ba51 : SUCCESS in 19m 07s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a703577df7994dc383dce11b9a03ce0c : SUCCESS in 6m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/18241a8e3b09451a948eaa78c95ff5f3 : SUCCESS in 7m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/db60f46c4e2044338f52cc77b37970b8 : SUCCESS in 11m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0e6bb64b3c8649d5a43e4e5927ddd130 : SUCCESS in 51m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e10d8e29393e41ecb20f186627660214 : SUCCESS in 1h 25m 48s","accounts_in_message":[],"_revision_number":1},{"id":"3c1b1e08f99f34c975007dff27686407ea64ae24","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 15:36:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"896e9d7c1cb9832771ce27e090063f9cf1ca11e8","revisions":{"896e9d7c1cb9832771ce27e090063f9cf1ca11e8":{"kind":"REWORK","_number":1,"created":"2021-07-22 10:53:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/68/801768/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/68/801768/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/801768/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/801768/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/801768/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/68/801768/1"}}},"commit":{"parents":[{"commit":"19e7cb43ea0b6e91c3b34a15dd2ea8ff0738075e","subject":"Merge \"update constraint for python-octaviaclient to new release 2.2.1\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/19e7cb43ea0b6e91c3b34a15dd2ea8ff0738075e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-22 10:53:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-22 10:53:25.000000000","tz":0},"subject":"update constraint for os-net-config to new release 13.2.1","message":"update constraint for os-net-config to new release 13.2.1\n\nmeta: version: 13.2.1\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: I9495cb36a9ac8f8dcb225c2c5839ab7b53d341de\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Icbc085ca48de9d53c10c502c3980d4dfe0db0e7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/896e9d7c1cb9832771ce27e090063f9cf1ca11e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/896e9d7c1cb9832771ce27e090063f9cf1ca11e8"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
