)]}'
{"id":"openstack%2Frequirements~801769","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~I6bc4c6756b1fe8fcf8f61bcb4fb2133677204660","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"I6bc4c6756b1fe8fcf8f61bcb4fb2133677204660","subject":"update constraint for tripleo-common to new release 13.2.2","status":"MERGED","created":"2021-07-22 10:57:16.000000000","updated":"2021-07-22 15:36:09.000000000","submitted":"2021-07-22 15:36:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"801769-new-release","meta_rev_id":"3aed5a24a7b6e3bf7c39fb9bc2b05ff327eeff3c","_number":801769,"virtual_id_number":801769,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-22 15:36:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-07-22 13:58:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-07-22 13:58:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-22 11:22:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-22 12:32:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-22 13:58:49.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"d1b8a2ea227b9b76b583a271dad8795a65a25111","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-22 10:57:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9610ae1131533f2b13335214cc0f28f7a16e8f16","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 11:22:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/42d0b7799e754ea6adc8dad547e4e42c : SUCCESS in 23m 31s","accounts_in_message":[],"_revision_number":1},{"id":"f8d4d7eb8096a2850682ac4de1dccb441dabb3c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 12:32:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba718dab5b10498ea602625784d5fcf0 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6fd11257d70f4e79af44b903976bce3f : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11487af79d9144ee88b5756609de8d51 : SUCCESS in 5m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/98fe64fc76194ab0a0486250d75245bb : SUCCESS in 5m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5a3ecf0874c24886a83df5a5de5422b5 : SUCCESS in 5m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/aeae6467d6ea407a96e52d093083b093 : SUCCESS in 9m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/836ab4234ef74141b1d5f9981ac9119f : SUCCESS in 8m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/39df686d38f24a3b82bdef8920eb54ea : SUCCESS in 9m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/441356304e2f4317833eac1fe429eec9 : SUCCESS in 5m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2af717d7717b41e5b8ae3e1c0ac2b5cc : SUCCESS in 9m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/45f9da3753c943228736763f0ec9086c : SUCCESS in 10m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3ae5fbcd504142cfbf4464fb5ce685a2 : SUCCESS in 8m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e82fb0d8d1dc49e6bd47a1ca6e94ad1e : SUCCESS in 15m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d5c896a855e24d1f8091aebed217ff80 : SUCCESS in 20m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0be900e64143469cac4c5de7ce94eb0a : SUCCESS in 13m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6f57be5a58304b1c98a90f23a67d9fec : SUCCESS in 4m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f922d1787eaf439e8e84596013e775bc : SUCCESS in 18m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e34ac6dce6574b28b782db846b913e43 : SUCCESS in 6m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9b11caa1f0fe4e6eb9bf400973b7f567 : SUCCESS in 6m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/74f0178c28f0481b8c0bd6b51f0bed82 : SUCCESS in 10m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18e4f4aae251477992be4b55705caf8c : SUCCESS in 53m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f00cc68e25b34af8b3ee03a61b8dfa8b : SUCCESS in 1h 31m 09s","accounts_in_message":[],"_revision_number":1},{"id":"7f0e080cdc83aca7f2083d15d1db154cb7f73608","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-22 13:58:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"00989c2de063866551f8d1d6d62e515fd59f2b37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 13:59:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f1c7f58123122f8011880211c947b404d5f44351","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 15:36:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20e11df3500c4272ac4ddfb49a2dac18 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e87ff9a3d63f48008b52c2748034ad2a : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03244939547e4192a5c14792de3de334 : SUCCESS in 5m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b5a1a8ecc2bd4a99a7007ef0c61e26ff : SUCCESS in 5m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ac7371eece99427891a897561dfd1001 : SUCCESS in 4m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8712f05c9bd740929936a900e108a783 : SUCCESS in 9m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d90b20d51443454a9823a809c10c0f0b : SUCCESS in 14m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/47a722d58baf4ebcba66756aefde2fdb : SUCCESS in 10m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ecfbd1a8b842492fae7677a87c74373f : SUCCESS in 7m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/01cff6c010eb4cdbbee26b5892341160 : SUCCESS in 9m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ce5eed6fa83540e9a99fcdc1d94ee0a3 : SUCCESS in 11m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/855760526b0a42cbaef7fd0fce987b77 : SUCCESS in 6m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2b662eccb3d7433fb6529e0f3b877049 : SUCCESS in 19m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ee09c30effa04692bf153815dc8d0c5d : SUCCESS in 23m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/30e6ad99db014e128cd8e4157754a236 : SUCCESS in 14m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1c913a16be214aeb94944a07ee7e699d : SUCCESS in 5m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5eb3705b6fd749acaba2f1705eb55f87 : SUCCESS in 17m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/40854ab6e56945e8ab2848a0b5a24f01 : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4ebce306a9c74fefa7a251af5d1ac041 : SUCCESS in 8m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4129c70f20a548ebafd11780dbcf0a20 : SUCCESS in 11m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2bd1ea651f31456f804b681c891abaa6 : SUCCESS in 52m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f5318eda606b465db81485dc8304d78e : SUCCESS in 1h 25m 56s","accounts_in_message":[],"_revision_number":1},{"id":"3aed5a24a7b6e3bf7c39fb9bc2b05ff327eeff3c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 15:36:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"20977f3f649d854eb6f944fb85cfd96f3d3db588","revisions":{"20977f3f649d854eb6f944fb85cfd96f3d3db588":{"kind":"REWORK","_number":1,"created":"2021-07-22 10:57:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/69/801769/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/69/801769/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/801769/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/801769/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/801769/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/69/801769/1"}}},"commit":{"parents":[{"commit":"19e7cb43ea0b6e91c3b34a15dd2ea8ff0738075e","subject":"Merge \"update constraint for python-octaviaclient to new release 2.2.1\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/19e7cb43ea0b6e91c3b34a15dd2ea8ff0738075e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-22 10:57:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-22 10:57:13.000000000","tz":0},"subject":"update constraint for tripleo-common to new release 13.2.2","message":"update constraint for tripleo-common to new release 13.2.2\n\nmeta: version: 13.2.2\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: I9495cb36a9ac8f8dcb225c2c5839ab7b53d341de\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I6bc4c6756b1fe8fcf8f61bcb4fb2133677204660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/20977f3f649d854eb6f944fb85cfd96f3d3db588"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/20977f3f649d854eb6f944fb85cfd96f3d3db588"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
