)]}'
{"id":"openstack%2Frequirements~802746","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I21566c8cd587deaef929c684c2b16d7ec8511cd0","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"I21566c8cd587deaef929c684c2b16d7ec8511cd0","subject":"update constraint for tripleo-common to new release 12.4.6","status":"MERGED","created":"2021-07-28 15:57:25.000000000","updated":"2021-07-29 03:56:25.000000000","submitted":"2021-07-29 03:56:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802746-new-release","meta_rev_id":"9e0a50ef1847e477d2dc6fa07e19df17d5d8c6b2","_number":802746,"virtual_id_number":802746,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-29 03:56:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-07-28 19:54:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-07-29 02:32:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-07-29 02:32:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-28 17:33:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-28 19:54:41.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-07-29 02:32:52.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"129dc9b19030d0a837325ec061683e0bc0c0eb95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-28 15:57:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ce327ca12373ca9cf883a9bb9755905e4313ad2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 17:33:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f767b2a4e6ef414a873b29d23c08c8d7 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4916526d42d7491b906e01e04b97351f : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b4d9ba135e943ae94399abd206e6a6f : SUCCESS in 5m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/db965385cf2848f7ad360dd32728d849 : SUCCESS in 4m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/63289b48ad0a4594942338cd39dc04bf : SUCCESS in 4m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5515b3bca4dd424988c3f9bc8aa6e16b : SUCCESS in 9m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5fe682517478427b936b922a2cce239e : SUCCESS in 9m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1ba9ce88d7cc449eab3f45dac9bbb9fb : SUCCESS in 6m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/18ecdee9bc3b4b88972acb7e8be383ac : SUCCESS in 10m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/58bc059ad0cc488bbbfd905caee84069 : SUCCESS in 9m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/063d33c9a5de45f6bcfcba560a3fc960 : SUCCESS in 17m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/235f1e1085ea4b2e8eadc4231bb0aca6 : SUCCESS in 21m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1256d2137dbe4739a248f17189d9cf2b : SUCCESS in 13m 07s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6b18f189e9fa472cbf73b85cd0201e85 : SUCCESS in 14m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/178cf6a94fbf4abeb6c066da5842f27e : SUCCESS in 17m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ad9183c5b65a4160a5232948b0c42a97 : SUCCESS in 6m 01s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/bac11b9fd08e43cc886094d717dcd47c : SUCCESS in 7m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/261bd8fe46c64f1eaba259ad2b1d1bec : SUCCESS in 7m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b749f7733c4a45539211a1f57c53883e : SUCCESS in 12m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b2b7a7a9238e45e798fa01d04ff7ef51 : SUCCESS in 12m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56c52aa085fc4efd9f97f6095b8c3dd5 : SUCCESS in 49m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b0d923a4b80847618e5bbdecf76d8e00 : SUCCESS in 1h 10m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/07634215202b4551a2c9ab6734152f26 : SUCCESS in 1h 21m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/97498d46bbd444cb8d11808b7d6bc111 : SUCCESS in 1h 29m 45s","accounts_in_message":[],"_revision_number":1},{"id":"4a7e32363d95da6560f133cefc13c23966f21624","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-07-28 19:54:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a097609ed8a6bdc0539399025a59959608523709","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-29 02:32:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f47d7bb243c6163bdadbda07069dec4b00fb3e4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 02:33:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"66d223e499a1761ee9f06962eb7653e6738ac938","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 03:56:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5f895dc6a244f22ac89739727139428 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/965c2c4f9f274799965197817805ab4c : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a00c583ea38941b9b4b77d6cba7a1c82 : SUCCESS in 5m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e212758932134bddb5fe3be9cac9cfdf : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a00141f5e8f2407e8159d347ffc2ce21 : SUCCESS in 3m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/837b55526f004b42ad203915ecc392af : SUCCESS in 13m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/79523d55a38d48a390a34a01f72f1197 : SUCCESS in 11m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8f563d0a15ca48b390035f2cae3e0d4b : SUCCESS in 5m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2cc59133dde64fea8493b47c7b6749e6 : SUCCESS in 12m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f2305517a9a44df2bd9c373f1abf0bea : SUCCESS in 13m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/289bf5b2df5641e4ba50b7628dbe4967 : SUCCESS in 18m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e41e8319ba1f4854be75470678414ae7 : SUCCESS in 25m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2dfc946927124635b2885313c43f5d82 : SUCCESS in 13m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a2b7849d9bc1443e90f5a9cae2556c04 : SUCCESS in 15m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/32e0d605034a46d0953d1052384d3b08 : SUCCESS in 18m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ef165a106ba04aea8feca76d3337bbcc : SUCCESS in 5m 25s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c9c2bf4d9516409da08fdf6a5697275c : SUCCESS in 5m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dfd68bdee31646d6a6cb350505f48ff5 : SUCCESS in 6m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bf85733a3743423898f8a7b9eec6e83d : SUCCESS in 11m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/27b522699ea14bd087cb9b776f2bf47a : SUCCESS in 12m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/652e9851a1ad4468a67c43d09b7a030f : SUCCESS in 53m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/77dbfb783dac4907ab32dbd5527aedc6 : SUCCESS in 1h 10m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3b9bd817f05a432ca246c02c350a3fb6 : SUCCESS in 1h 22m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/7ca68f423cc94bf68dbd9962caaa0539 : SUCCESS in 1h 11m 22s","accounts_in_message":[],"_revision_number":1},{"id":"9e0a50ef1847e477d2dc6fa07e19df17d5d8c6b2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 03:56:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"033c5da7768203a55a61451ebf2c9a1f6f8e555c","revisions":{"033c5da7768203a55a61451ebf2c9a1f6f8e555c":{"kind":"REWORK","_number":1,"created":"2021-07-28 15:57:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/802746/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/802746/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/802746/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/802746/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/802746/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/802746/1"}}},"commit":{"parents":[{"commit":"3685bd3aa1eb60f0d8b70e313e063d79509036eb","subject":"update constraint for ceilometer to new release 14.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3685bd3aa1eb60f0d8b70e313e063d79509036eb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-28 15:57:23.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-28 15:57:23.000000000","tz":0},"subject":"update constraint for tripleo-common to new release 12.4.6","message":"update constraint for tripleo-common to new release 12.4.6\n\nmeta: version: 12.4.6\nmeta: diff-start: -\nmeta: series: ussuri\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: Ied1c09bf9f6d22d9de7728ad4dfdff89e73f9ae0\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I21566c8cd587deaef929c684c2b16d7ec8511cd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/033c5da7768203a55a61451ebf2c9a1f6f8e555c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/033c5da7768203a55a61451ebf2c9a1f6f8e555c"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
