)]}'
{"id":"openstack%2Frequirements~802748","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I9ad5ceb7db5346ea37120313ef97afc89fcf5f65","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"I9ad5ceb7db5346ea37120313ef97afc89fcf5f65","subject":"update constraint for os-net-config to new release 12.3.5","status":"MERGED","created":"2021-07-28 16:02:57.000000000","updated":"2021-07-29 03:56:27.000000000","submitted":"2021-07-29 03:56:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802748-new-release","meta_rev_id":"2e084f0f352cea4a4599d01213cb99b78813ed81","_number":802748,"virtual_id_number":802748,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-29 03:56:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-07-28 19:54:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-07-29 02:32:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-07-29 02:32:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-28 17:42:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-28 19:54:33.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-07-29 02:32:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"a74e33e9ee27438cfff1a379b5a2d3b954d2f719","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-28 16:02:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"adf13fb69a353bd393242f1d8495b66641b9b3c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 17:42:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/224ffc6445d84d1fb8ad0e5d4977d5b4 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72f7fd693c304c89b95b7fc585f8ac6c : SUCCESS in 6m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/66c921d16d604ce48b0383d9fd08df5e : SUCCESS in 6m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0d59b271d1b34ceda435ffa4cadf2e6c : SUCCESS in 6m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7a1c87748ae2469b9f71319c4b14cdfa : SUCCESS in 5m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/44a12406c13b4da0bffdd3bda1d10ac1 : SUCCESS in 10m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/86a9144ddf5441eb911cc149e864b1cf : SUCCESS in 11m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5428049f0a334271bd5e4a3b9b628a18 : SUCCESS in 5m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ff36e6c007e14e0fa9dc82e0166423be : SUCCESS in 12m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eeb70a4e783d45c1a92724f11017bfc5 : SUCCESS in 8m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/aaa8c279496e409c9044f9413d9181b4 : SUCCESS in 13m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7cb47e2cd286401d93e4756bb20dd2d4 : SUCCESS in 25m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7c2882f13a3f4955abdcd94dfb9f85e5 : SUCCESS in 13m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7ebcdb97e49c4b1dbcd72339adb899a1 : SUCCESS in 12m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/378f8ed7a6434ea4b489365b883f2c18 : SUCCESS in 17m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/319abb2822fd45358e9c7f82457c2297 : SUCCESS in 6m 21s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/cd0deb3ff27d4d5a9d3e127ba59725c3 : SUCCESS in 5m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/173266159f824cf0865757b1f3eddbb2 : SUCCESS in 7m 09s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/48aa8740767343baacc9ec08495d942a : SUCCESS in 9m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/72950944edfc449bbe42fc7593d1f4e6 : SUCCESS in 11m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1566e6b7fed3421dbfd1a1fb70bde6ae : SUCCESS in 49m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb3b9cb3c9294e22961292f1a393c4f5 : SUCCESS in 1h 27m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/61a6877931e44c3aa1b7e117fb5cc15e : SUCCESS in 1h 38m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/17fac9cece9740e2abc1f7b1b0203d9f : SUCCESS in 1h 18m 17s","accounts_in_message":[],"_revision_number":1},{"id":"26d10748392442360d0020ecf70502fdaa06ed4b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-07-28 19:54:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3a9026812ce26aa5ac65008848b46d7230adc6a8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-29 02:32:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4f0b35b0d376866ae9b99f70aa36046cc1ba3a67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 02:33:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c97b6fa508717cd5b9ab80109bdbfdf4de5300bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 03:56:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9ed70380279482e9063a90e91ceac47 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afd96ab6732743adb4b8f538db799f6c : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf702b7f440a4d78b5813928fc60edeb : SUCCESS in 4m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c3c8dfd1506b403496e9606bb2945a64 : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/00f1debc258847bd924ea1222d558022 : SUCCESS in 4m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0cb9467b8e284970bb512c64a66edddb : SUCCESS in 10m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1f4da320ac2a4765b936c6361c9966e3 : SUCCESS in 9m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8a1ce7e836374a658fbb996efa3eeb52 : SUCCESS in 5m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/381af478b5324729b8358d9723cde4fc : SUCCESS in 11m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3b4e36e7f2cf466e8f21c6d041c76e31 : SUCCESS in 8m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cbbd7624ab84453f8341f181d0407a53 : SUCCESS in 12m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b7c062d3236f4b26b5e2b860830e6bd7 : SUCCESS in 25m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/531c809cebaa4a46a04e835be773d595 : SUCCESS in 13m 26s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/aae6304fcd3b4c2aa68cbf50ffb03533 : SUCCESS in 12m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8706cbee2afb48ec992c63c1f7d7a834 : SUCCESS in 15m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/470b67f5c97e4982b0341ed8587c56e0 : SUCCESS in 5m 13s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/8c814c5c7bdd4a3aa094b13657593a7f : SUCCESS in 6m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/90bd3dc82f9d4c21924bece275fe35fc : SUCCESS in 6m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/47c545c3b140422d98e5f9167e7ef233 : SUCCESS in 10m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/86f8e9b5496b4994adf90328a478e0c9 : SUCCESS in 11m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3abdf96f2a01471aa55df099f3ba5a94 : SUCCESS in 50m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7000b63860034c2da681297938d7d66f : SUCCESS in 1h 08m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c2ea7192cd4e46ee86f77d7cc77d8add : SUCCESS in 1h 16m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/2b2dda7f082449d7a86303d83bf64c76 : SUCCESS in 1h 07m 26s","accounts_in_message":[],"_revision_number":1},{"id":"2e084f0f352cea4a4599d01213cb99b78813ed81","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 03:56:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0b08d4322c2f63babae219b9d10bf21386974cff","revisions":{"0b08d4322c2f63babae219b9d10bf21386974cff":{"kind":"REWORK","_number":1,"created":"2021-07-28 16:02:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/802748/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/802748/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/802748/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/802748/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/802748/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/802748/1"}}},"commit":{"parents":[{"commit":"3685bd3aa1eb60f0d8b70e313e063d79509036eb","subject":"update constraint for ceilometer to new release 14.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3685bd3aa1eb60f0d8b70e313e063d79509036eb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-28 16:02:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-28 16:02:55.000000000","tz":0},"subject":"update constraint for os-net-config to new release 12.3.5","message":"update constraint for os-net-config to new release 12.3.5\n\nmeta: version: 12.3.5\nmeta: diff-start: -\nmeta: series: ussuri\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: Ied1c09bf9f6d22d9de7728ad4dfdff89e73f9ae0\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I9ad5ceb7db5346ea37120313ef97afc89fcf5f65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b08d4322c2f63babae219b9d10bf21386974cff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b08d4322c2f63babae219b9d10bf21386974cff"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
