)]}'
{"id":"openstack%2Frequirements~802828","triplet_id":"openstack%2Frequirements~master~I213a130e57b61ac5f0d1afab7ca42fef6e754a34","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I213a130e57b61ac5f0d1afab7ca42fef6e754a34","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2021-07-29 06:26:53.000000000","updated":"2021-07-31 17:48:47.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a4f9deb60e29e94aafc4b7f6361063e5d264618e","_number":802828,"virtual_id_number":802828,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-07-31 08:01:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-29 06:45:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-29 08:12:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd569e1836b9e50ab7fcb6eaf1a6354ac981cbdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-29 06:26:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57efe2479d2bcc0c69a660db4eb053c186b26380","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 06:45:41.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2cfa5ef0809e499fa8fc9bf40d6fd76e : FAILURE in 13m 58s","accounts_in_message":[],"_revision_number":1},{"id":"b2a37c879b37df6e30d8184a1aad15c1a1f5d68a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 08:12:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/803d4bca42ca4372bbdaa2d60128c30a : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f306167858fb4d20a2736ba154a8792d : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f27503ac5bc94a12975c32b8aa821607 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c146a5b0e51c42b580bebe7862cfbd06 : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0342663ff7b84393ac4e868ccd7332fa : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/61eb0d57a2bc4e96adcc9784fc5d5351 : FAILURE in 6m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0af0bfd5887644aca3c9516f136df82f : SUCCESS in 6m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e605631ca6464342b481e72515669821 : FAILURE in 7m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5ed107b44f1442c296dafbe7bab1314d : FAILURE in 5m 34s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/3b8840f9fdbd4b0d90e679e8175361d6 : SUCCESS in 6m 00s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/59cace08776d4bd496b4f4c6be464106 : SUCCESS in 9m 52s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b63aa7e26c3c486ca98034e2d002d673 : FAILURE in 6m 00s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/eecf5bfad3d54fa491987405e3ad2e3e : SUCCESS in 10m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7e27a25ca9744d96a56655c51383d75c : FAILURE in 7m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7f07936e43e74186aa72fccae9612d96 : SUCCESS in 8m 50s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e7fc9dd51fa344a88eb5598deb9e44ff : SUCCESS in 18m 40s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a46c00d6b6304514bc0d88495d3dcd70 : SUCCESS in 11m 30s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ee6d8e30464941a695ef2b9440351784 : SUCCESS in 5m 23s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/fe83cf7c713846a4afbedcf6156f59ad : SUCCESS in 23m 49s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b3c6082cd765488ab429eb964d4d9f7e : FAILURE in 14m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5b1253e8df2f40b083f2ea70d5d6427e : SUCCESS in 4m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/465117d25ff6485685069c31914afbce : FAILURE in 18m 39s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/8c4c57d7fe3e490c9eda4692a175d0d9 : SUCCESS in 6m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/69d006e27d264971a0f3d7f9223d5a4c : SUCCESS in 7m 54s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d88385ef0965491883d734f718802dde : SUCCESS in 12m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a572e6870c04b3a974afaf6186b7f17 : SUCCESS in 44m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/22623527b18548a9a332f6f87310a37e : SUCCESS in 1h 17m 40s","accounts_in_message":[],"_revision_number":1},{"id":"8539dabf135c053fe060492020c8eba76442ba07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-30 06:29:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1c1f788f1dcef2f12c7d77933df5671ca15a7cff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 06:46:41.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ffc58d28b426477bab572c02b3b65790 : FAILURE in 11m 18s","accounts_in_message":[],"_revision_number":2},{"id":"8f4cc4da62323a1c1ed8ab3dd9aee98ff232dc95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 07:55:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbad649adbac477e86eae86858e13da6 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4088a5a98fea4ab29b4c78cacf51a298 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80e685d978a241f1b30f16497ffe826d : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04ec1b5623c14076a382272b098bd3dc : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5deb84e73a5c4d76b4a7e789bd86a04c : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d4472e7df7f448629c3327691ff6461d : FAILURE in 4m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a9495c1c2b8a478080aea499c771bd1b : SUCCESS in 5m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ef046b8c6bd54161bbd850513f17f794 : FAILURE in 6m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b19068187e82402c95fd09bda84cd0b3 : FAILURE in 5m 35s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/54a84dcac3dd47f78168d1a305680608 : SUCCESS in 7m 29s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/61b47c776e0a4a6bbfc05db9cc9f6684 : SUCCESS in 10m 18s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ef84583fd92641bda58aafa39d591a94 : FAILURE in 7m 43s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ce3cb6b90e534c938583ce5da6c8145f : SUCCESS in 12m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7b63bb2c3ebc4adfbf538fbc9773f1bf : FAILURE in 10m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/75c41ed0787a4a899acf054e65782bac : SUCCESS in 12m 02s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6a1a58a4a40e402a9db1279f95ab833f : SUCCESS in 17m 10s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2f99fb42a400445f89cb4879817f849a : SUCCESS in 10m 42s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/cef4345e252b4538bd1457260309a8e1 : SUCCESS in 5m 32s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6893c53be6744ac9b5287b3a0c60ca88 : SUCCESS in 23m 55s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/eb99b82ad1134069a9a359f750c0dcfd : FAILURE in 13m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6ae781a145f44a66af72e55633ff7e59 : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/19a488f83f9740629c6b8e36b54d4b6f : FAILURE in 18m 39s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1fe853bbd58f47e480e6577a7d6bd153 : SUCCESS in 6m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/38bc0ff7dbe94d0b8f4af23ae6255b8f : SUCCESS in 8m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7500cc7f9fed435a9928d38c40b50993 : SUCCESS in 14m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da3ef3e11c8a48fcb1e61f1efbf7076a : SUCCESS in 46m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3074a3049191452084bd8b227906e705 : SUCCESS in 1h 14m 52s","accounts_in_message":[],"_revision_number":2},{"id":"b0282c343d9f032f2c1e4312032922411e89ed2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-31 06:34:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fb95a0aea5adee15521a618db11ef137d548cbf8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-31 06:51:08.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/210c5cabd393437aab957993196fc6d6 : FAILURE in 11m 28s","accounts_in_message":[],"_revision_number":3},{"id":"cb33627714e281aaf629e55ab4fcf05389448ee0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-31 08:01:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0de40fc8cff4e868f09c8247257ca11 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e9d03bcdc704366b904e8640ebe9423 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec429e9fb5a74f128b943d04cdb183d3 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0f3a46ad98c4fe496fac41fae3cd0a5 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba962e942f6941b093902254971c3eb9 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4b4e39501e6b4ef490e92102dfe0e92a : FAILURE in 4m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b4a9526ae2634ad5a6766492771879dd : SUCCESS in 6m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fe5dcb66320647a7ba895998b671ffa0 : FAILURE in 5m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5bbfac7ae27a41c083a88df3974a0111 : FAILURE in 5m 41s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/fa96a772441e4b8e991b41fd1b4eeaa7 : SUCCESS in 7m 47s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/df3a2edf0e5d49b8a2cd2e3c9537788d : SUCCESS in 10m 18s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0c06f55697e04afd8a549c887ac90680 : FAILURE in 7m 28s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4a077c055b12483d90180cbd48e21367 : SUCCESS in 11m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2e9327b471264177a06dbae0ecf73ede : FAILURE in 10m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/54f0d045d6ee4ef2b196157d58b9d29e : SUCCESS in 13m 01s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a4fa89d811ad421bb63c7a256f208403 : SUCCESS in 20m 04s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/86da6bca20304b8a8ef36ca595be1ff1 : SUCCESS in 12m 59s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/701f85dccc9a45b8b530b31972e22ed8 : SUCCESS in 8m 54s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/23a1201ea20a4422bdb7e74f8484abbb : SUCCESS in 24m 12s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/16db61fc672e4e918b19ce7d0c1b65b0 : FAILURE in 15m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d502c20b143140bd989aee20343b9362 : SUCCESS in 5m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/70b4c3c9743047d981a5b7844f9b72b8 : FAILURE in 18m 25s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/51b32f889808496c8fad14f104e22c83 : SUCCESS in 5m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/537b2d3e599f4b6a8fb820aa82ec6730 : SUCCESS in 8m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/bca9120ad2764a8abcc359fb06d8a805 : SUCCESS in 13m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0225bb88c5640a8b323e02bed847b6c : SUCCESS in 51m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8411cfe1dcd3403db8897e8cb53475f6 : SUCCESS in 1h 20m 39s","accounts_in_message":[],"_revision_number":3},{"id":"a4f9deb60e29e94aafc4b7f6361063e5d264618e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-31 17:48:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7701e2d53691b11eab9547299a84a3382c7811ac","revisions":{"2551b10feff9f79930adfe19830b627ef428b0ca":{"kind":"REWORK","_number":1,"created":"2021-07-29 06:26:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/802828/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/28/802828/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/802828/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/802828/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/802828/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/28/802828/1"}}},"commit":{"parents":[{"commit":"9eea28c09aa73e5abd9a862050dbf5b250e90848","subject":"Merge \"update constraint for monasca-common to new release 3.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9eea28c09aa73e5abd9a862050dbf5b250e90848"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-29 06:26:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-29 06:26:51.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I213a130e57b61ac5f0d1afab7ca42fef6e754a34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2551b10feff9f79930adfe19830b627ef428b0ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2551b10feff9f79930adfe19830b627ef428b0ca"}]},"branch":"refs/heads/master"},"55ca32d5341821c721348cf531f3e771d40b7156":{"kind":"REWORK","_number":2,"created":"2021-07-30 06:29:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/802828/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/28/802828/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/802828/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/802828/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/802828/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/28/802828/2"}}},"commit":{"parents":[{"commit":"27c057840b74687729484b01847d978cf8b7f444","subject":"Merge \"update constraint for tempest to new release 28.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/27c057840b74687729484b01847d978cf8b7f444"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-30 06:29:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-30 06:29:53.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I213a130e57b61ac5f0d1afab7ca42fef6e754a34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55ca32d5341821c721348cf531f3e771d40b7156"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55ca32d5341821c721348cf531f3e771d40b7156"}]},"branch":"refs/heads/master"},"7701e2d53691b11eab9547299a84a3382c7811ac":{"kind":"REWORK","_number":3,"created":"2021-07-31 06:34:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/802828/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/28/802828/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/802828/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/802828/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/802828/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/28/802828/3"}}},"commit":{"parents":[{"commit":"27c057840b74687729484b01847d978cf8b7f444","subject":"Merge \"update constraint for tempest to new release 28.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/27c057840b74687729484b01847d978cf8b7f444"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-31 06:34:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-31 06:34:53.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I213a130e57b61ac5f0d1afab7ca42fef6e754a34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7701e2d53691b11eab9547299a84a3382c7811ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7701e2d53691b11eab9547299a84a3382c7811ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
