)]}'
{"id":"openstack%2Frequirements~802852","triplet_id":"openstack%2Frequirements~master~I9166cfa1e80c2fd4537720d86bed5bbb49e90e21","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I9166cfa1e80c2fd4537720d86bed5bbb49e90e21","subject":"update constraint for horizon to new release 20.0.0","status":"MERGED","created":"2021-07-29 09:26:00.000000000","updated":"2021-07-29 17:56:04.000000000","submitted":"2021-07-29 17:54:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802852-new-release","meta_rev_id":"3a297d59ca6ec6883d47a057151d99f543ca33f1","_number":802852,"virtual_id_number":802852,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-29 17:54:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-07-29 15:05:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-07-29 09:53:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2021-07-29 10:35:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-07-29 15:05:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-29 09:44:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-29 09:53:41.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-07-29 10:35:27.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2021-07-29 11:13:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-29 15:05:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"dcf6d389c6c304fb88155fb846e1999779def0ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-29 09:26:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"63dd2518a831ee6901e1dde617a98240b62d6283","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 09:44:59.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7e779b4ab05248cfaae83fae4ba3c1e0 : FAILURE in 16m 55s","accounts_in_message":[],"_revision_number":1},{"id":"e4e5d8666e2c38339562f60bb5550dcd415abd4e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-07-29 09:53:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b42511403afa0dd94f1e94efad998f25fb0210ce","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-29 10:35:27.000000000","message":"Patch Set 1: Code-Review+1\n\nwe(trove-dashboard) need this.","accounts_in_message":[],"_revision_number":1},{"id":"d2efa8200e297dc23bede22fbad63f8bbae6bec0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 11:13:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/870f6ac5f0a343239a5d3a31836a0b23 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5063c54bbc5e4c5fb3ea8361fdd65e7c : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4789c21da62d43efb83f46d98c64459a : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7f43fa80e1246f5b80c6e7810b51830 : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e11e916e423a41eda9f6dce294736105 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6bf909ad45b74a448b9b10081b97860e : SUCCESS in 4m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/783894d465c34faa958fd53b216dca1f : SUCCESS in 4m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/64d6f4850cc54bdbb0ad0c1784b7dd2c : SUCCESS in 10m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/28f335e1c5f94ad9958d799aac41ccfd : SUCCESS in 9m 20s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/db0e2747d8e743bcbc056e763d619a8c : SUCCESS in 6m 40s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/487fce0c6aa34bc2a84d0b4571ab18f6 : SUCCESS in 9m 15s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7b2f25f7a9b94fb9a7f1aef662890f94 : SUCCESS in 5m 51s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3506924b968d483286c620612a0d0454 : SUCCESS in 9m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/903977e6fb4c4a2b816a1081f9e8de5e : SUCCESS in 8m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d2a1f19f51f04120bb19e49dd7a1f490 : SUCCESS in 10m 41s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d2fb99f82a694b8594c4ed5b74178170 : SUCCESS in 19m 04s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e0fe194340a449498240daf6b0140f57 : SUCCESS in 9m 18s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5f380d833f94477caaf5deb125aff539 : SUCCESS in 5m 43s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d686aa11866b4ded821df120e0d72d26 : SUCCESS in 28m 56s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/daeda2cedfb84618bcb5701531e7ddc6 : SUCCESS in 14m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/746a6ea30e3e404186d06600d851e5e0 : SUCCESS in 4m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7a1116399d864518ac4452fc69b40181 : SUCCESS in 18m 05s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7d8dce65607c4e3e9ac1e568fda72944 : SUCCESS in 6m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f84176e551f8452aaad6e478365a19ea : SUCCESS in 9m 09s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2f458111aae34f08a5b981392f250569 : SUCCESS in 13m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d720b66252734b9ba5d3c5b38bbfb742 : SUCCESS in 50m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0de5b594c87f4e199315bf8a65a0d9b3 : SUCCESS in 1h 20m 14s","accounts_in_message":[],"_revision_number":1},{"id":"a1294c835585daaed1273734644994d29fb0f7fa","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-29 15:05:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"83be114fb53ca474c4dd1988a0ecc036abdd4517","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 15:10:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e432a94a6c6cedf4ac875dc57ea39eb669482d6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 17:54:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73e2550490504e0487df58720ae7535a : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e794de355f84a168f0590f1b60d8107 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f4ad38f9b7440a5be2690f873b1b3d4 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/966052f9cc684e39bf14829b74c5c42a : SUCCESS in 7m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3e3a8522292f47f893312c01034ef6d5 : SUCCESS in 5m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/17366b3948bf434ab576e86947572884 : SUCCESS in 5m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d69be3205dff4ba888173fb80c1ae2e1 : SUCCESS in 10m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/31e375e5d0a448d38c0a520715e16c73 : SUCCESS in 8m 59s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5f592de976134bc8824881f9f7938c3b : SUCCESS in 11m 10s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a1ec0de5eacb497ba77f81e356990de5 : SUCCESS in 6m 35s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0fd01610f16e4b8a83a131312cb0ca25 : SUCCESS in 10m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0572a6e0ee8940b1a9afaec93c848fd1 : SUCCESS in 8m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6a7b0c3554bd47a5970f252e328f0e8e : SUCCESS in 10m 48s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/cad2d1ba8fc544458465a7d83e70f964 : SUCCESS in 14m 31s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/490c8688209e48db8f3717662b351a44 : SUCCESS in 12m 31s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/da73b12336c64fb49965bc6fbefc9469 : SUCCESS in 5m 53s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a0cc402c423945f3b146e1f0bc1196a9 : SUCCESS in 24m 35s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a3d86eff1753484c8167a5e802e333f3 : SUCCESS in 14m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/77172ef5037541eda85afc3885ebabf1 : SUCCESS in 4m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4fdade4679c84b8a89b1bc70b9e829f8 : SUCCESS in 20m 36s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/99071869e5fc4c6cb834c192efd13ea9 : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fa1cfd7307eb45058d84b2aef3a1eed5 : SUCCESS in 7m 58s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e3345aba85ee43c7ae6666ecbdf78ae0 : SUCCESS in 13m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83de521249164caa948f64e94d9515c2 : SUCCESS in 42m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/102cd8fb8d5f46f5bf65fc89785b310c : SUCCESS in 1h 12m 52s","accounts_in_message":[],"_revision_number":1},{"id":"7430bfce62306f4d376075d207145715a5c080d9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 17:54:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3a297d59ca6ec6883d47a057151d99f543ca33f1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 17:56:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e729c674e8047fdbb4615870a4a0ba6 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f74c6dedba1930d1d01e7697ee2d0b9b7088954b","revisions":{"f74c6dedba1930d1d01e7697ee2d0b9b7088954b":{"kind":"REWORK","_number":1,"created":"2021-07-29 09:26:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/802852/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/802852/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/802852/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/802852/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/802852/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/802852/1"}}},"commit":{"parents":[{"commit":"9eea28c09aa73e5abd9a862050dbf5b250e90848","subject":"Merge \"update constraint for monasca-common to new release 3.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9eea28c09aa73e5abd9a862050dbf5b250e90848"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-29 09:25:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-29 09:25:59.000000000","tz":0},"subject":"update constraint for horizon to new release 20.0.0","message":"update constraint for horizon to new release 20.0.0\n\nmeta: version: 20.0.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: manchandavishal \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Commit: manchandavishal \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Change-Id: I392217ea8d18b1347a71b3f782350280b234a3db\nmeta: release:Code-Review+1: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Itsik Kanias \u003cikanias@redhat.com\u003e\nmeta: release:Code-Review+1: Lingxian Kong \u003canlin.kong@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I9166cfa1e80c2fd4537720d86bed5bbb49e90e21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f74c6dedba1930d1d01e7697ee2d0b9b7088954b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f74c6dedba1930d1d01e7697ee2d0b9b7088954b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
