)]}'
{"id":"openstack%2Frequirements~802855","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I05a95c89459f088c5ad5bd2b7bd94e2d9404c872","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"I05a95c89459f088c5ad5bd2b7bd94e2d9404c872","subject":"update constraint for python-zaqarclient to new release 1.13.2","status":"MERGED","created":"2021-07-29 10:08:04.000000000","updated":"2021-07-29 20:51:09.000000000","submitted":"2021-07-29 20:51:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802855-new-release","meta_rev_id":"2b018d3bd0ab2a181cf37b69916c31471b63d102","_number":802855,"virtual_id_number":802855,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-29 20:51:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-07-29 19:27:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-07-29 10:16:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-07-29 19:27:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-29 10:16:40.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-07-29 11:59:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-29 15:06:01.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-07-29 19:27:06.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"339faa5e648583443849c2cd78a4b3ff567d9d0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-29 10:08:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec891304eaaaa483eb9a807957786676c3e95b41","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-07-29 10:16:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"da11935635e31233b182eeb19a61647e03c963f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 11:59:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8c32e1cfbe44e23941bc50b0d612afe : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a77b7e11ba1481e98194babdecaff3a : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4210b799c64544608ea683bcef47a7d7 : SUCCESS in 5m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e2f4834eee254db19d32d75537cc0a58 : SUCCESS in 5m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/741d53c4eb754f1aa9d0ab871394e1c8 : SUCCESS in 4m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c62568221747449b924a6f077b51c4a6 : SUCCESS in 12m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/38ce6e5cf9cb41709079a3a5fa72d507 : SUCCESS in 10m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e9adfe9bd61042e2930bac8abc41b3b6 : SUCCESS in 7m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3fb181ce973b4c6db54e274f407a8c5a : SUCCESS in 10m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fd761236a58247d3ac901f0150195064 : SUCCESS in 9m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0d0eb2034f594b7985e194a41436b268 : SUCCESS in 15m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1a9450aa6ee2495d90b0b08e8ec5c702 : SUCCESS in 22m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1de7b76d84fa4e718ecffb7a278b89e6 : SUCCESS in 13m 54s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e8a81e8074894a52b4a00d3f2c51f162 : SUCCESS in 14m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c02f0dca313c43b3998b12065c5fab60 : SUCCESS in 18m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/996ad513a9e74737ae25419a1ce0a640 : SUCCESS in 5m 28s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3882d8aa525944bfaa42fd2ceaa56869 : SUCCESS in 6m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5ca22c4a3c13457eb3d9f74cca04a551 : FAILURE in 4m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/24ba66989a5e470ea3f58c0ccfdf92f7 : SUCCESS in 12m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fba75c7c728740fba52dbdee1e353ae5 : SUCCESS in 12m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c86c246f4235422a8bab7213c8f3894d : SUCCESS in 51m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/497c6195f7e04bb0a8542b41bbd1a59d : SUCCESS in 1h 17m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fff4f5dd19874b988d96c87669f022c9 : SUCCESS in 1h 42m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/6798c707893e4a588b4c9fc184dcdd48 : SUCCESS in 1h 10m 14s","accounts_in_message":[],"_revision_number":1},{"id":"bc811f6a0fdbfbbe07ad679cf0f11c669eebfcc9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-29 15:06:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d03bc2bdce807ba727db78ec6980262626824335","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 16:50:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a50ae8dd3c8468dbe8ec42ccc343053 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27cfc6e2866b448db51832aeeb0e60e8 : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2ba6608405794361a8a537e2465f7f59 : SUCCESS in 5m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/08b832c8e6894958a4f8739646c38de0 : SUCCESS in 4m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/92950080b26d4994b821c4dd77b179b5 : SUCCESS in 5m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f0c6efb3c9c041b7ad541e231b69fca2 : SUCCESS in 9m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7e5c96454648496a9e9b592e6cc38ead : SUCCESS in 9m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/da7ba231dc9a4b99bdde11c4beb4512e : SUCCESS in 5m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6762beed62f04138b1503fe1c11037ae : SUCCESS in 11m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/04f5e60b6083432291e4159d43aa3a38 : SUCCESS in 8m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/80bb2fb4d5964a5cae98590743e14037 : SUCCESS in 19m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/86c41fde842d415a9e39aff7aa8561db : SUCCESS in 22m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a5aded7882384ff7bb25b7da37d2cebe : SUCCESS in 13m 27s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/112ff8c67ccf4e51abe70269543a3be4 : SUCCESS in 17m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7fe40da68c1b4400b6a35a2333fe48ca : SUCCESS in 16m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/97160c30b6e7487e8ff7eecd721e4ae2 : SUCCESS in 5m 48s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/58290aed3caf43c4bdf6939b1d539990 : SUCCESS in 6m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4882a60b3cbd4090a5b3deb581373538 : SUCCESS in 7m 18s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/275f990e6ffa470dbe5402687cbd5194 : SUCCESS in 12m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6aa7c2b23a4149c587aa8c505b8b7706 : SUCCESS in 11m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10b1f0fde69d4020af0c6e51c6bfa4c6 : SUCCESS in 53m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a4a12bd271f74d16ba70c037b369fc5c : SUCCESS in 1h 10m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/157c467b035140beb40093849ae9b324 : SUCCESS in 1h 34m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/aeab93c5cc0a4816afa8b4908ea9a2dc : SUCCESS in 1h 15m 36s","accounts_in_message":[],"_revision_number":1},{"id":"3fa8927e13420145ca70662cc380583bd0e27ee6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-29 19:27:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4833395e1e39e1cc9069faf6e29f7308eb129356","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 19:27:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e4b5aa7c1377c5567e5b024954ad5ccacfb83172","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 20:51:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26895968f4924f1b89f50f997478cdf3 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b133e70588f04217b506da34dd60b866 : SUCCESS in 6m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32e915f79bdb40c7873a286da106a968 : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7294ebd00ac646dcb293a0c244fb0796 : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/65090710caec41799aed9c7d35c1daf9 : SUCCESS in 3m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ee3206ad7270473ea1164cbf41b393e2 : SUCCESS in 9m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4b02cd4180dc41a2ba8468c27f2389da : SUCCESS in 9m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f5c620a7afab45cab4cf629b595049b3 : SUCCESS in 6m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5d68c5b75855486f90e633cf034ad363 : SUCCESS in 10m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/25a2a7284777493696680e1cf12a0fd2 : SUCCESS in 8m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ebb1ee67f6e74503b44650f103d6d582 : SUCCESS in 14m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d185cb51c9884313a952019222967f9a : SUCCESS in 22m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/62a5b280813d4b6c965ec6f24da5b34a : SUCCESS in 12m 08s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0cde33ceffa143d5a49be3d5812f2d40 : SUCCESS in 13m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2245e0b56e3a4103a73bc2bedb5a58ca : SUCCESS in 16m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6129c8c1fece48ab91589878dfb6f529 : SUCCESS in 5m 41s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/46e1c2a2f64e42e3b79776dfd54220b8 : SUCCESS in 7m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bb5fc7c638884afaab1f285a02a43e95 : SUCCESS in 6m 26s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/972c87b999664d54955d79720252aebf : SUCCESS in 11m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a22b26498fad4803a28199425d239cef : SUCCESS in 12m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/128028077b504ab2b9f460363fc65cc9 : SUCCESS in 45m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/feb42b32963f489da8cf80125c63dfa8 : SUCCESS in 1h 12m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c8c557d943d54fcda6eebcef2b4ac61a : SUCCESS in 1h 21m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/df49c94333044d7991742c693bfdc722 : SUCCESS in 1h 07m 05s","accounts_in_message":[],"_revision_number":1},{"id":"2b018d3bd0ab2a181cf37b69916c31471b63d102","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 20:51:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2941c615dcc440b7b11467c755e8dc19ea95b7d","revisions":{"a2941c615dcc440b7b11467c755e8dc19ea95b7d":{"kind":"REWORK","_number":1,"created":"2021-07-29 10:08:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/55/802855/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/55/802855/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/802855/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/802855/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/802855/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/55/802855/1"}}},"commit":{"parents":[{"commit":"6f03ec365fc7027204812d7364bd17a702dd20cf","subject":"Merge \"update constraint for os-net-config to new release 12.3.5\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f03ec365fc7027204812d7364bd17a702dd20cf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-29 10:08:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-29 10:08:03.000000000","tz":0},"subject":"update constraint for python-zaqarclient to new release 1.13.2","message":"update constraint for python-zaqarclient to new release 1.13.2\n\nmeta: version: 1.13.2\nmeta: diff-start: -\nmeta: series: ussuri\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ib36c5d1ab52c5a4276121bfa611ccb64c956b549\nmeta: release:Code-Review+1: 王昊 \u003csxmatch1986@gmail.com\u003e\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I05a95c89459f088c5ad5bd2b7bd94e2d9404c872\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a2941c615dcc440b7b11467c755e8dc19ea95b7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a2941c615dcc440b7b11467c755e8dc19ea95b7d"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
