)]}'
{"id":"openstack%2Frequirements~802856","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I9db20038befc3cfb2731a031d28fa17ca32e1bac","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"I9db20038befc3cfb2731a031d28fa17ca32e1bac","subject":"update constraint for python-tackerclient to new release 1.1.1","status":"MERGED","created":"2021-07-29 10:15:13.000000000","updated":"2021-07-29 18:21:35.000000000","submitted":"2021-07-29 18:21:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802856-new-release","meta_rev_id":"1528614e45af34e8351c59d4c803e8f0193a7d48","_number":802856,"virtual_id_number":802856,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-29 18:21:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-07-29 15:05:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-07-29 10:16:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-07-29 15:05:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-29 10:16:52.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-07-29 11:42:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-29 15:05:54.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"8bd8c2946bfd016fa5fe530080758b81c1fa4e5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-07-29 10:15:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb28d45a960f2acb1a5b04bedfa111b57c7c31ef","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-07-29 10:16:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f74d7960f412ad64f1516e546f81925adfd7fb90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 11:42:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60396089a3e1447b81b36eb73e1bf06b : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e84f0a6092c4d2692efafe84e9c057e : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0a8411379bdd4d0a81fda940494791b4 : SUCCESS in 5m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/76b1249348a3460eb41e705a0dc8bd40 : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c56dd867e2364506ba5df1e572344eb1 : SUCCESS in 4m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f02c81c248f9427697b06cf63187ab4c : SUCCESS in 9m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/13801864c9024724a9b38bd9ff08fd81 : SUCCESS in 9m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a1631f12165042bd9502073e9b952204 : SUCCESS in 6m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ef92f7ee85614694876ccb0d127878fd : SUCCESS in 11m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/43fd9d168c3642b6a4742587c53c8d56 : SUCCESS in 10m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/185cb42695744aaeb2067681a7301369 : SUCCESS in 14m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/941a2cadc12d4e2ca5be6f6b68f44fa0 : SUCCESS in 22m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ee2d1119edcc4028972748a74bf9aec8 : SUCCESS in 13m 21s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/61936a9b52c34490bfda570b5ca4d9bb : SUCCESS in 13m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b8ce1f0a81f04a9cb835cc98fdcd5e6d : SUCCESS in 15m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0f239c3059dd4f38917d11bd14cffa61 : SUCCESS in 6m 11s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/1f0f2cbace434a0fa8c09bab9f54ea18 : SUCCESS in 8m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/570d4e9d3fcd442281b289f1529268c9 : SUCCESS in 7m 22s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e12d0fb7d7dd4a76afaa2d68558a08cf : SUCCESS in 9m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a498099b4989414e837c726e6bb11f0b : SUCCESS in 13m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/759673dc427940ce9e28c1daec07f11f : SUCCESS in 48m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fee73e4d8a7a4ddb9ba2dec91c8f9e8c : SUCCESS in 1h 10m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/47a52aac41774143ba3b34d6ad1022e2 : SUCCESS in 1h 26m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/9b183ef148c8497c8c61da93c2d64a06 : SUCCESS in 1h 04m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d429a6d0edcdbf0a4af4788a08f5dc097f4d9b5b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-07-29 15:05:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0351d8f99c253f2a02798609c994e8627208414b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 15:10:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"40b214999376739f98e0f2894ec304fbc21d8c3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 18:21:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4660ef0fcf524959a5b0d83f6166af1e : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8380f04fc7224fbaa55377d4669dc8e4 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/407a99d3d5ed446b9f0a422c819ca77f : SUCCESS in 5m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e286e0a2033546e7af5d2f905d3e05a4 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/035ed0a525f143a8b16877e76e9a22b7 : SUCCESS in 4m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c5924d74d5d24078817951e611539a45 : SUCCESS in 11m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2e23d8ba8dc647479408805f6d9eba22 : SUCCESS in 10m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d12a3962d9394d98a8c7756504d87171 : SUCCESS in 5m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/271ff55b8c78439598599615cd74cee7 : SUCCESS in 11m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/01f4e06a2248419cbd00f0e1bf05a8a7 : SUCCESS in 9m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/79978a88f19e4c6fa94472278f93ba49 : SUCCESS in 14m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1cd731a38af144009c00740f4430860a : SUCCESS in 26m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6ea52a374f524d92867375401611033b : SUCCESS in 14m 10s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0900d6d7bbc541628b7caac9de4185c0 : SUCCESS in 13m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0d83cd150101491494de33529906c44c : SUCCESS in 20m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8be03a0229264da699f42aa3d8ad7fb1 : SUCCESS in 6m 02s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/7c48dba2deba435aba5e6c8b2c2a9fe7 : SUCCESS in 6m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8927ac13c54248fca1a29f920e22c766 : SUCCESS in 7m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/98ff9938bd5c4b08a0666b367b89fd62 : SUCCESS in 11m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0a9259658018425d861083a82d4dac71 : SUCCESS in 11m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/97803c4406264e5486d7404547ae1e23 : SUCCESS in 58m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0da5d76f8ae4480e87d73b0e2470e8ed : SUCCESS in 1h 13m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7b48a7cd9b354ae3a8ed18d154843e4b : SUCCESS in 1h 35m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/7033d76a8038410fa53e1831e853caa7 : SUCCESS in 1h 21m 38s","accounts_in_message":[],"_revision_number":1},{"id":"1528614e45af34e8351c59d4c803e8f0193a7d48","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 18:21:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"907f01f6be474d1783ef45485dd1baf1eb3e5f10","revisions":{"907f01f6be474d1783ef45485dd1baf1eb3e5f10":{"kind":"REWORK","_number":1,"created":"2021-07-29 10:15:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/56/802856/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/56/802856/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/802856/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/802856/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/802856/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/56/802856/1"}}},"commit":{"parents":[{"commit":"6f03ec365fc7027204812d7364bd17a702dd20cf","subject":"Merge \"update constraint for os-net-config to new release 12.3.5\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f03ec365fc7027204812d7364bd17a702dd20cf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-29 10:15:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-07-29 10:15:12.000000000","tz":0},"subject":"update constraint for python-tackerclient to new release 1.1.1","message":"update constraint for python-tackerclient to new release 1.1.1\n\nmeta: version: 1.1.1\nmeta: diff-start: -\nmeta: series: ussuri\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I68e718e93c51c9e4111f4a8a2072e218fe04884a\nmeta: release:Code-Review+1: Yasufumi Ogawa \u003cyasufum.o@gmail.com\u003e\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I9db20038befc3cfb2731a031d28fa17ca32e1bac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/907f01f6be474d1783ef45485dd1baf1eb3e5f10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/907f01f6be474d1783ef45485dd1baf1eb3e5f10"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
