)]}'
{"id":"openstack%2Frequirements~803724","triplet_id":"openstack%2Frequirements~stable%2Frocky~I077e6276618a257b2c948e58d4de993998844827","project":"openstack/requirements","branch":"stable/rocky","hashtags":[],"change_id":"I077e6276618a257b2c948e58d4de993998844827","subject":"[stable-only] Use bionic for requirements-check job","status":"MERGED","created":"2021-08-06 10:30:44.000000000","updated":"2021-08-06 20:39:17.000000000","submitted":"2021-08-06 20:39:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"803724","meta_rev_id":"8351c267fe0d2b1f7d20967ba07ea2611a26777c","_number":803724,"virtual_id_number":803724,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-06 20:39:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-08-06 11:52:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-08-06 15:13:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-08-06 15:13:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-06 11:52:01.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-08-06 11:58:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-06 15:13:35.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"729ed69b4427cf5c2831deecf68edaedd5aa0476","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-08-06 10:30:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ca6032253a245564de5e032083fdfa42e3b007f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-08-06 10:33:44.000000000","message":"Patch Set 1:\n\nsee job failures: https://zuul.opendev.org/t/openstack/builds?job_name\u003drequirements-check\u0026branch\u003dstable%2Frocky","accounts_in_message":[],"_revision_number":1},{"id":"4a1bf3d41befd3f34f145d18fc39d021f1b0af87","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-08-06 11:52:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f1d6ee4d6a9b7da053a925801de07c8619615f33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 11:58:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5db20452413e4c5cb9095faa6c3b7ebc : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e98e9e6e66344b81a372ee8a2bb8725d : SUCCESS in 5m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a8d8732ec41c456098b6495da9c471a2 : SUCCESS in 5m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b8c8bd57b17e496db694b137c2a018b4 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4023fcaa1bbf4072adaa21289267d5f0 : SUCCESS in 4m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3c5d590a92784c889f2816d57578d766 : SUCCESS in 1h 10m 40s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/f2deaa7ae68e4ad4a45391d184df4fda : SUCCESS in 1h 20m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b6b9a5125d864083bf45a8c5a4aac33a : SUCCESS in 1h 01m 28s","accounts_in_message":[],"_revision_number":1},{"id":"2adfaa1a952e8d9c6164f4b526b11f2c55ec4382","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-06 15:13:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bedc907a04f93e79e32e9953cc7783795c98641a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 15:14:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c1bf114adfb3c285072a6d246a6f7e4b292d71e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 20:39:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/222afb41ec2841da9cc404fe3db457e2 : SUCCESS in 4m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/82f74150035b437ab5be268dd454a8d9 : SUCCESS in 10m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7ca7c18d82bf46dfa56fce780b37edc9 : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/286538bc8cde4aa6a9e9a7ad7cabf46f : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c15cfe75c3644ea8ba69dec62db15bd1 : SUCCESS in 5m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ca781e140c2e4e499b5ffe5265f71a3c : SUCCESS in 57m 58s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/2b607d8b621c43319d10381c3c37a1e5 : SUCCESS in 1h 35m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/66281192d4b6479c8104862fc20f0327 : SUCCESS in 49m 11s","accounts_in_message":[],"_revision_number":1},{"id":"8351c267fe0d2b1f7d20967ba07ea2611a26777c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 20:39:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"725f494f307bd986cb68835dcfa7a4eedf1a8043","revisions":{"725f494f307bd986cb68835dcfa7a4eedf1a8043":{"kind":"REWORK","_number":1,"created":"2021-08-06 10:30:44.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/24/803724/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/24/803724/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/803724/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/803724/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/803724/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/24/803724/1"}}},"commit":{"parents":[{"commit":"adb8d5da24a4c5e9b1087f6588616c43961bbcc6","subject":"[stable-only] Cap hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/adb8d5da24a4c5e9b1087f6588616c43961bbcc6"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-07-21 15:02:07.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-08-06 10:27:59.000000000","tz":120},"subject":"[stable-only] Use bionic for requirements-check job","message":"[stable-only] Use bionic for requirements-check job\n\nrequirements-check jobs started to fail in stable/stein with\n\n  ModuleNotFoundError: No module named \u0027openstack_requirements\u0027\n\nThis is due to that job runs in ubuntu-focal instead of ubuntu-bionic,\nand openstack_requirements is not installed for python 3.8. By pinning\nthe nodeset to bionic we can eliminate this issue as in bionic the\npython version is 3.6.\n\nNOTE(elod.illes): older stable branches fail, too. In rocky and older\nbranches Xenial was the supported nodeset, however the check job will\npass with Bionic, no need to use the older distro.\n\nChange-Id: I077e6276618a257b2c948e58d4de993998844827\n(cherry picked from commit 3a3ed8a7d36c2321a519fd6b54b7b328274e4d07)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/725f494f307bd986cb68835dcfa7a4eedf1a8043"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/725f494f307bd986cb68835dcfa7a4eedf1a8043"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
