)]}'
{"id":"openstack%2Frequirements~804484","triplet_id":"openstack%2Frequirements~master~I1aefbc12c6cbc967a295131b0344612cc5c8c664","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I1aefbc12c6cbc967a295131b0344612cc5c8c664","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-08-13 06:33:09.000000000","updated":"2021-08-13 18:59:08.000000000","submitted":"2021-08-13 18:57:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804484-openstack/requirements/constraints/noclob","meta_rev_id":"c8b06395719689826793bc19f59c69ab58a6c919","_number":804484,"virtual_id_number":804484,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-13 18:57:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-08-13 17:22:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-08-13 17:22:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-13 06:50:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-13 07:58:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-13 15:37:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"0631ee79708190f8f012124c89bae32be593d2ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-08-13 06:33:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e2e789da5aacb69d86c9a82e04fff4d2e751c46","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 06:50:09.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8effdb95424348289bd8ea380a57eda7 : FAILURE in 11m 27s","accounts_in_message":[],"_revision_number":1},{"id":"79b8b24725cb5b5a3c4b954e0ef1fb0a83301fbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 07:58:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68fd8326463d45089e5ed30397bac0fc : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e24ad53925049acb67fac032fd34449 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/433f96c8ddcb461396e1640883f5f915 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4dc5e910a6f4b65aa47fb3d45ff25bb : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/718376d0456f416183fd0d911eb0319a : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7e159ff851584e3894123c1e18e29c39 : FAILURE in 4m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/18bde3644ce346099fec2faa55eb695e : SUCCESS in 4m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e3e2744e1d924385a3670f42e0ccda2e : FAILURE in 5m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/46928bbac6b94bb0af8bfde21246c24d : FAILURE in 4m 51s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e109573e44994852b0a4ce3be0bac249 : SUCCESS in 8m 02s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ab578034cc3547d2a9b843e4ea134e5e : SUCCESS in 11m 20s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4b9fb6eb6d6c49c3808d9250050625d4 : SUCCESS in 9m 15s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/522411599d8543dc9dc2b6ba0d7daf9c : SUCCESS in 11m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ce9949845c114b6a880599b94f4eb12d : FAILURE in 9m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/10a0ec48647e48b6ad753a2bd38134c8 : SUCCESS in 10m 27s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8681fca82feb494e9261338f89ced147 : SUCCESS in 20m 26s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/617a34fa53ae47eda45f8a6cde04eb72 : SUCCESS in 12m 48s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/acf5614758bb4583a6fabad25199b212 : SUCCESS in 7m 16s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/61257d2fbd5f4234b08d12e5766ef516 : SUCCESS in 22m 35s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/45b23e0698744c4881f759514b9c0ccb : FAILURE in 13m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/43440461b1994c39a895c438fc69c14d : SUCCESS in 4m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7c3052ce0bb84c669118c77b005429a2 : FAILURE in 21m 54s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/29d42019bdb749ffb4e32a8cfacc5532 : SUCCESS in 6m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/589947186cea4679a74ac899eef44a8c : SUCCESS in 8m 21s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cb79a3f9290c40a6bb3221e04e231f07 : SUCCESS in 12m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/86a43948de0746369d27de0e68658914 : SUCCESS in 46m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/056ce8af5a2743c7bf9c0faea87cdd27 : SUCCESS in 1h 15m 22s","accounts_in_message":[],"_revision_number":1},{"id":"f8916addffdddff1c416060c1b6bce1687c1d2ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-13 15:37:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dbacd5ef377783c81093cc2bf98160e41e1522f2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-13 15:37:51.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"10b91b269ded6f6bd200c5e0d488fe2334578b62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 15:54:16.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/18bc365e404f46e69ba543251c59758d : FAILURE in 12m 04s","accounts_in_message":[],"_revision_number":2},{"id":"47ffb04744e350d3b753fd43c3c22bd2b88ba70a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 17:02:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a29f8146886f4292a4d2d18af4b64659 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc99f443ea9b4437b10ba07defb92b7f : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/718d44729afe46b684864167ac46e428 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/060f9bc1fc02444ebb2c761048823d5d : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8dff938b87ec41ffab4558a18c22c976 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ab2d09fff7a647929bb01b0e518e8751 : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b8a0bc12f65245d1aba897ee1f60abfe : SUCCESS in 4m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c7660d73251849c786377f0c79edd29f : SUCCESS in 10m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ad0e26010d4248fda487516e0ec91313 : SUCCESS in 9m 53s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e97280b3cbb74b7fa048ca0fb5494047 : SUCCESS in 5m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/92c693279ab640388bcee4f08a873858 : SUCCESS in 12m 14s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ab703d6a4f414b11aee7a13ca587ccfb : SUCCESS in 6m 37s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3b344739d15f48a293b88c4936438a19 : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c315b60d6f6f47d397b25afce9b7fd9b : SUCCESS in 10m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/66eb420c53194d788212209f9bd17a08 : SUCCESS in 8m 07s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/91cc65a25c064cbf97b15bc5fc1f79ba : SUCCESS in 15m 06s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ce815308c4134500be089b14182c5c85 : SUCCESS in 8m 59s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/172d48808f2c4847b54c206799ff9a60 : SUCCESS in 8m 14s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4c0a6b7975ea487da9bd31eb75a2e279 : SUCCESS in 23m 34s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/818ad116367b4900ba6c8596101da196 : SUCCESS in 13m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bcdb2474c386477892dd6dbd87279b4b : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0210a6b23fb146f6a8ee3f8676a13ba3 : SUCCESS in 17m 33s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/8c96c1f945f84bdaae0b800402f4317e : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/701aed17ac5049aa86117d10d9745555 : SUCCESS in 7m 45s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/31b5f23700664cc7816d8f8685ac01a5 : SUCCESS in 12m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa5e669cd321423f9503986065d559a8 : SUCCESS in 44m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/041af4d90ea7432990101d1efe427c09 : SUCCESS in 1h 19m 36s","accounts_in_message":[],"_revision_number":2},{"id":"14831df4e092516964ad9655511b0ddbe186f477","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-13 17:22:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"74023df03413d743a3dec23dd8e01cafb8f0e06f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 17:22:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cdf3f1256d5a817111a06d88e57c4b93658dea88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 18:57:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2230d681f3ac451c983010592fe6a653 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e7b32acde274b408a9203526cf91616 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/936a2123c47d4c51a14d0b55102d4135 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3369b38acba460eaf5e645e237e452f : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/41f6611f65a14ab792430f156e87b938 : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c954fa3253264d81b57258ded30ae803 : SUCCESS in 3m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fb0fe1d0e5aa48bf87d60e30e8219037 : SUCCESS in 11m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0a771cf2cf784d3d93882e80bb9e2ad0 : SUCCESS in 8m 36s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/1ee735e0de8d426cb73230cc8b9c7d0a : SUCCESS in 8m 17s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/1325e9a201a64e879be33c81cf3649b4 : SUCCESS in 5m 15s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1549764f3e334fa8a33cc8ef6db37fc4 : SUCCESS in 10m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b11eeeb957dd4c678c73ad9aa5195e9c : SUCCESS in 11m 02s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e22a3dc35a8b4b30ba66c03a7da182b8 : SUCCESS in 9m 01s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/3ef865ffb5a54d9e9201eff3606444f3 : SUCCESS in 13m 52s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f23ae98c158e4b34a3c8f3d98e86d8b3 : SUCCESS in 8m 44s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/24f46c4891c145a99b838e749a83708e : SUCCESS in 6m 22s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e0914777b4a5422099cfd95c5b49f0a1 : SUCCESS in 20m 53s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/20c180d09c0a4b7baf051198d7fc2edc : SUCCESS in 15m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7ea34dce679d44288b0db184d721c5c6 : SUCCESS in 3m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eedc317712e6414483154fa92728fd74 : SUCCESS in 15m 49s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/68a24610fa1c4ce199489cf354a4124d : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/698b021312d44f6eac5028f300e8f878 : SUCCESS in 6m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/57b2151820664781aa285c757b989bf7 : SUCCESS in 11m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/76e6b4632ce841f8a900e3c16eb38032 : SUCCESS in 45m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eb55b9bc6d1247f68e937db6e84a64ab : SUCCESS in 1h 30m 37s","accounts_in_message":[],"_revision_number":2},{"id":"29f248d826e1896c996f5719e192299a730df765","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 18:57:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c8b06395719689826793bc19f59c69ab58a6c919","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 18:59:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4691bbf5348348b79d9da03612e5e296 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"40b0fea7ea2e358bad0c26e897761403959aa3d4","revisions":{"a7255b40062e501f4914b8a6c236f1bf3a03cbdb":{"kind":"REWORK","_number":1,"created":"2021-08-13 06:33:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/804484/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/804484/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/804484/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/804484/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/804484/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/804484/1"}}},"commit":{"parents":[{"commit":"39fa27685556fad19595e51b29abcd6446cbd88e","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/39fa27685556fad19595e51b29abcd6446cbd88e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-13 06:33:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-13 06:33:08.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1aefbc12c6cbc967a295131b0344612cc5c8c664\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a7255b40062e501f4914b8a6c236f1bf3a03cbdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a7255b40062e501f4914b8a6c236f1bf3a03cbdb"}]},"branch":"refs/heads/master"},"40b0fea7ea2e358bad0c26e897761403959aa3d4":{"kind":"REWORK","_number":2,"created":"2021-08-13 15:37:42.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/84/804484/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/804484/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/804484/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/804484/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/804484/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/804484/2"}}},"commit":{"parents":[{"commit":"39fa27685556fad19595e51b29abcd6446cbd88e","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/39fa27685556fad19595e51b29abcd6446cbd88e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-13 06:33:08.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-08-13 15:37:35.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1aefbc12c6cbc967a295131b0344612cc5c8c664\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/40b0fea7ea2e358bad0c26e897761403959aa3d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/40b0fea7ea2e358bad0c26e897761403959aa3d4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
